)]}'
{"id":"x%2Fpackstack~731464","triplet_id":"x%2Fpackstack~stable%2Fstein~I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2","project":"x/packstack","branch":"stable/stein","hashtags":[],"change_id":"I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2","subject":"Fix CI jobs","status":"MERGED","created":"2020-05-28 08:34:32.000000000","updated":"2020-05-28 14:16:14.000000000","submitted":"2020-05-28 14:16:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731464-1590675374919-06556da8","meta_rev_id":"1b94ad7b5be9038a449c82ec647326c5fa1cd25e","_number":731464,"virtual_id_number":731464,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 14:16:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-05-28 11:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-05-28 10:59:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2020-05-28 12:11:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 10:59:25.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-05-28 12:11:07.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-05-28 14:16:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26922a72108f035ad57b9e367a7c3b8c5169522b","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-05-28 08:34:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1603967da5c182718f04a38c6d9a83c2e9d0cbe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 08:57:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c2ee10400c342dea8ed23c8a743843a : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e114c9f8deb74c09aaa21f32c3cd3161 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bed292d505a94450bd8f3835f022bc45 : SUCCESS in 3m 47s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/94e5afb45f814aed8682303b7b70b917 : POST_FAILURE in 15m 47s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4ca14ff4060247a0a77478febe695b9d : POST_FAILURE in 17m 53s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/53b0984924fd45f2a0d634560ff1be90 : POST_FAILURE in 16m 18s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d3b6b353a9ae4fb9a7b39831f381e7df : POST_FAILURE in 14m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0721d21f48074b9badd1eba36cdb8b5c : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fd741aa3757e7245800d7d33187f21a468977b7","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-05-28 09:36:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9bcf1b3f8ff405920afafde69a84c2e5d17a7c6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 10:39:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8254a6bd84254d06a10660cf5bf30e1b : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab9eb8d22f624b329fbbb1c222e0fce2 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/623bfceb18c846928630e0af462758cd : SUCCESS in 3m 56s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/8ea32460a41240c997d67481df7fd734 : SUCCESS in 55m 13s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2a3429c68289458eaa7cbce2227b9442 : SUCCESS in 47m 09s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/8755dda00c624881ad0e4c1e0890cdb0 : SUCCESS in 57m 25s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a0b8d7e591244241b099a4956d60c5c4 : SUCCESS in 51m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12b3301b6e2742028378db2890636451 : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4bed586c4b70fd0d7c9200365a1fa1eaef7f8d8c","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-05-28 10:59:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1841d703f8d992b2756399ca34e040e92aaeccb5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-28 11:04:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"55ada9759e8478d735bb1e54c5964eddec6df565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:04:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9842eb6e4d2b47565227a599a8e78e03fe1cad2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:04:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16e42bd055b0419097e123afe80b5ea0 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76b3ec6f38104071a3848c45a135d664 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57d7bae150eb4f08a5567f2488312b7c : SUCCESS in 3m 44s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/0279d62b661a42c99e6b20055e144e50 : POST_FAILURE in 9m 19s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9dd51dc4a6a943ba8ec78f4b03f96998 : SUCCESS in 51m 51s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/05385f34f218499eaa7172eaec2ecc78 : SUCCESS in 53m 10s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6516f8ae9e974b0dadb7ca120649e14d : SUCCESS in 46m 31s","accounts_in_message":[],"_revision_number":2},{"id":"8139fff2ac82a6be780b34e82a9ef6233c952985","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-28 12:11:01.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"d002882cbaa2d17e923ecd9ded64a189577b1767","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-28 12:11:07.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ece78a6861bad0340420f7603be234ba942376ba","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-28 12:12:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1b9cbda6391ecaf14acd974e038d9ba570a3727a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:02:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/185b427b29e342aaa8a010a5800fb84d : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96b38399cda64ac68ba73bc794298154 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f94a4e6525749a0a37b8af63afbf850 : SUCCESS in 4m 06s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/8bc02078cbaf48dabc3bc2221d5bed66 : SUCCESS in 47m 53s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/820e9ca3e9b644708f0943a75405de86 : SUCCESS in 40m 14s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/ceabd81d6b8747868a75d2a0d4344deb : SUCCESS in 44m 17s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c1cf0a6e77e24ce8992bb828b9449654 : SUCCESS in 47m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8944377c49c45f3b959a3f48cd4557b : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9816e36b4a1848ca71607c1655486c902cb42263","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:07:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63f427cd3e2544135adaa2a059ee1684187f3994","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:16:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e6ea7f878e5483db7813da2da800e55 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/921165ad77254692a0828cc3dd2d5deb : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cf57e52f4964739b9ddc558a18dc110 : SUCCESS in 5m 26s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/4b12268e1c18462ba1d0c622816d29e7 : SUCCESS in 56m 09s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/fc16aa0541cf4d1c8eb153a30b7547c5 : SUCCESS in 46m 30s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/14543d46626d4b75bc2faa7933b7d022 : SUCCESS in 53m 18s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3ae68b3349574551a68de2b5ec316a22 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":2},{"id":"6d52f93bbf355d9fe2c6db928f144026a43191ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:16:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e4ed593d649ad792a4a947064cf95b6ae9bd274","revisions":{"210d2e2847f12c214a3dd0a68ac7c494ee235cea":{"kind":"REWORK","_number":1,"created":"2020-05-28 08:34:32.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/64/731464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/64/731464/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/64/731464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/64/731464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/64/731464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/64/731464/1"}}},"commit":{"parents":[{"commit":"00bcd23a2795f0d59cb1113f5c8f100e16224636","subject":"Remove explicit compress of log files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/00bcd23a2795f0d59cb1113f5c8f100e16224636"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-26 14:31:19.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-28 08:34:19.000000000","tz":120},"subject":"Fix CI jobs","message":"Fix CI jobs\n\n- Add jobs removed in [1] to the local configuration.\n- Fix SSH key distribution in multinode job. It was broken on a\n  recent Zuul update, it seems.\n\n[1] - https://review.opendev.org/730813\n\nChange-Id: I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2\n(cherry picked from commit 76b1559d3a161b0d2326ada77bc69d089639075f)\n(cherry picked from commit 100c950bc7838d691841f8d0f2801eef9579ba5e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/210d2e2847f12c214a3dd0a68ac7c494ee235cea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/210d2e2847f12c214a3dd0a68ac7c494ee235cea"}]},"branch":"refs/heads/stable/stein"},"7e4ed593d649ad792a4a947064cf95b6ae9bd274":{"kind":"REWORK","_number":2,"created":"2020-05-28 09:36:55.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/64/731464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/64/731464/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/64/731464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/64/731464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/64/731464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/64/731464/2"}}},"commit":{"parents":[{"commit":"00bcd23a2795f0d59cb1113f5c8f100e16224636","subject":"Remove explicit compress of log files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/00bcd23a2795f0d59cb1113f5c8f100e16224636"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-26 14:31:19.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-28 09:36:30.000000000","tz":120},"subject":"Fix CI jobs","message":"Fix CI jobs\n\n- Add jobs removed in [1] to the local configuration.\n- Fix SSH key distribution in multinode job. It was broken on a\n  recent Zuul update, it seems.\n- Use pip by default instead of pip3 if present (needed for CentOS 7).\n\n[1] - https://review.opendev.org/730813\n\nChange-Id: I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2\n(cherry picked from commit 76b1559d3a161b0d2326ada77bc69d089639075f)\n(cherry picked from commit 100c950bc7838d691841f8d0f2801eef9579ba5e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7e4ed593d649ad792a4a947064cf95b6ae9bd274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7e4ed593d649ad792a4a947064cf95b6ae9bd274"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
