)]}'
{"id":"x%2Fpackstack~731465","triplet_id":"x%2Fpackstack~stable%2Frocky~I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2","project":"x/packstack","branch":"stable/rocky","hashtags":[],"change_id":"I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2","subject":"Fix CI jobs","status":"MERGED","created":"2020-05-28 08:34:55.000000000","updated":"2020-05-28 13:00:13.000000000","submitted":"2020-05-28 13:00:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731465-1590670813319-c44eae83","meta_rev_id":"33578f5c4386d5e49496ca79013d20ab79bea89f","_number":731465,"virtual_id_number":731465,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 13:00:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-28 09:34:53.000000000","_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-05-28 11:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-05-28 11:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2020-05-28 11:04:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 11:00:43.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-05-28 11:04:34.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-05-28 13:00:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0b09861b267e4768f88e7dd364877d838f7c300","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-05-28 08:34:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6896e8e7cbea20a59bf928be15a5e05f1e00c879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:29:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e96edd7f954fb893d77ac6c996effa : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77f88c13607d44e08d00945f18a0c96a : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/838a9dffd8204bb7a9110ad96b6da4c1 : SUCCESS in 3m 21s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/c31f6d8d0cc7467688dec809ce7bc457 : SUCCESS in 46m 17s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/faeec87047a146cfa4dbaf7f9b2bdae4 : SUCCESS in 44m 52s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/6719d4f9826f442db33ec8f0aafda875 : POST_FAILURE in 16m 41s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bb125eee1045417bb7f5c71adb865915 : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":1},{"id":"19177efe06f1570fa311f90e7db9067f42a3e50a","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-05-28 09:34:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8a25fd5e5f3c7ddc37bcbb19afa478c46b147b6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 10:34:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df3d98b32b2349d8a5c4956f7b7c9a18 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0771a7f8ac2942efb99e0271e4845ac9 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16488f04522f413384110162a75bd40f : SUCCESS in 4m 02s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/a07b230a14d74c4191933bac0096f51d : SUCCESS in 49m 52s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0ceb8020eb6846358ce58e1169a15980 : SUCCESS in 45m 18s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/b1095cde6bfd4066ab62d0d42160f98b : SUCCESS in 54m 12s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0a5611e972284c66a6b7d93e3663627e : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":1},{"id":"630d372659a616dcbbc036b1168d63bf06ca1fd1","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-05-28 11:00:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cbe29c3a5105b8d4f6494b8331956b07d6d7ecac","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-28 11:04:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f5871a521c60004103cc6d52ad58f123b557c87a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:04:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c19901b3a8b6b87d7f85fd0cc839022f729bb7ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:00:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"33578f5c4386d5e49496ca79013d20ab79bea89f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:00:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20559cfc36db4a25ab34c9b7d9705bda : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/260279ff5aed4175b30df0981cbd58ff : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a409a359d47f4fc28388bf46610a53b3 : SUCCESS in 4m 23s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/44a179d32f474b8e9a397dff987d28ac : SUCCESS in 39m 30s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/020e210f7dea4146a0423a0254fa38ec : SUCCESS in 40m 56s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/614743bfb54643cfb7ff482b6890eea6 : SUCCESS in 49m 05s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bb6fa9f8ae2545d6848c85349eb5e7c9 : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ee7479b85a5fa1ca93ef6330f1fcea3de687f45","revisions":{"3ee7479b85a5fa1ca93ef6330f1fcea3de687f45":{"kind":"REWORK","_number":1,"created":"2020-05-28 08:34:55.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/65/731465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/65/731465/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/65/731465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/65/731465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/65/731465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/65/731465/1"}}},"commit":{"parents":[{"commit":"58c8413dd8cf5eeab72c625b1f7faaf781058b88","subject":"Remove explicit compress of log files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/58c8413dd8cf5eeab72c625b1f7faaf781058b88"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-26 14:31:19.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-28 08:34:41.000000000","tz":120},"subject":"Fix CI jobs","message":"Fix CI jobs\n\n- Add jobs removed in [1] to the local configuration.\n- Fix SSH key distribution in multinode job. It was broken on a\n  recent Zuul update, it seems.\n\n[1] - https://review.opendev.org/730813\n\nChange-Id: I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2\n(cherry picked from commit 76b1559d3a161b0d2326ada77bc69d089639075f)\n(cherry picked from commit 100c950bc7838d691841f8d0f2801eef9579ba5e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3ee7479b85a5fa1ca93ef6330f1fcea3de687f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3ee7479b85a5fa1ca93ef6330f1fcea3de687f45"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
