)]}'
{"id":"x%2Fpackstack~731467","triplet_id":"x%2Fpackstack~stable%2Fqueens~I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2","project":"x/packstack","branch":"stable/queens","hashtags":[],"change_id":"I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2","subject":"Fix CI jobs","status":"MERGED","created":"2020-05-28 08:35:25.000000000","updated":"2020-05-28 13:00:14.000000000","submitted":"2020-05-28 13:00:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731467-1590670814119-3a602513","meta_rev_id":"92f7d739495e246b587cb519033e7bcaaeeb87ad","_number":731467,"virtual_id_number":731467,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 13:00:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-05-28 11:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-05-28 11:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2020-05-28 11:04:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 11:01:02.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-05-28 11:04:56.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-05-28 13:00:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93b0343b4e3b4cec7ca447dbb1237d97eea116f0","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-05-28 08:35:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f05f302a226835f0a5689b615266e2eaead7e51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:34:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aad9b0cfb3241fbb1c04b2ee68f86e0 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f6126aa73bb4466b40b91bd030163f3 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4b324eeb8ed47ff88ffa8311023c919 : SUCCESS in 4m 09s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/9968bb84955e4e6ba2205b60f42b05e9 : SUCCESS in 47m 25s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/07f7abbcf81b42ac81eb294904f5923c : SUCCESS in 45m 50s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/379f0da024bb468685475cc2bc1bf558 : SUCCESS in 53m 22s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3d0ed4fae591411ebfe4e51493e66753 : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e812d834eaef1d6f6dbfff19be0ccd275c30a13e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-05-28 11:01:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44bda9c76984221f3912d89bdedc9239131e0c7e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-28 11:04:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fcea31105c07e721f226f39e7b9089b57c1c88e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:05:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce666335855804fd83fded9e80adfa317a220e84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:00:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67f0d92f41d540aaa5f6ffcdf6284b96 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c0e701466f94548b848fefb3b85d45a : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/701b25307ba04ffd885f08a280ebf2c1 : SUCCESS in 3m 34s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/a5e509a24c04477da403e79ba10238b7 : SUCCESS in 48m 49s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/211e261d84464aaf99d113e1130db973 : SUCCESS in 48m 15s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/2354a6e9592c47f1b55a20b2fd3d0bae : SUCCESS in 54m 45s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/48d65bf853414591b67d85e8bd825a5b : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":1},{"id":"92f7d739495e246b587cb519033e7bcaaeeb87ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:00:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfffafa90606792e49bee49156bf451fce922148","revisions":{"cfffafa90606792e49bee49156bf451fce922148":{"kind":"REWORK","_number":1,"created":"2020-05-28 08:35:25.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/67/731467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/67/731467/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/67/731467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/67/731467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/67/731467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/67/731467/1"}}},"commit":{"parents":[{"commit":"4d4f78c2a1213f54fdd5ba72b67efe59bb9a425e","subject":"Remove explicit compress of log files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4d4f78c2a1213f54fdd5ba72b67efe59bb9a425e"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-26 14:31:19.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-05-28 08:35:04.000000000","tz":120},"subject":"Fix CI jobs","message":"Fix CI jobs\n\n- Add jobs removed in [1] to the local configuration.\n- Fix SSH key distribution in multinode job. It was broken on a\n  recent Zuul update, it seems.\n\n[1] - https://review.opendev.org/730813\n\nChange-Id: I2572d92cbaabc1f02fc0790896ccaa3edfc5e6c2\n(cherry picked from commit 76b1559d3a161b0d2326ada77bc69d089639075f)\n(cherry picked from commit 100c950bc7838d691841f8d0f2801eef9579ba5e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cfffafa90606792e49bee49156bf451fce922148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cfffafa90606792e49bee49156bf451fce922148"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
