)]}'
{"id":"x%2Fpackstack~736667","triplet_id":"x%2Fpackstack~master~Ifdccd8609658b3170b8356bddc5062f252677d40","project":"x/packstack","branch":"master","hashtags":[],"change_id":"Ifdccd8609658b3170b8356bddc5062f252677d40","subject":"Let apache module purge config","status":"MERGED","created":"2020-06-18 11:55:52.000000000","updated":"2020-06-18 14:02:35.000000000","submitted":"2020-06-18 14:02:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736667-1592488955728-8d5793de","meta_rev_id":"8fe5e48892ed01eed25c918f25fdbe7ffa7ca6d2","_number":736667,"virtual_id_number":736667,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 14:02:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-06-18 11:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-18 12:40:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-18 13:09:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 11:59:30.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-06-18 13:09:15.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-06-18 14:02:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3501742a63d4557516fd2fc3855f17ff9cf4a36","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-18 11:55:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a266abcb85a278e8aa851990c2bd8b60e13d9a3","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-18 11:59:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3f3e7da6e617337b5752bd4222d33394ed4722f9","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-18 12:40:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1faf7e29ad0e5b743855980839273dab71325a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:56:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed1831c265d44861a60afeb174042fde : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9e238d26d394e1cb1be47dc87d77dc5 : SUCCESS in 4m 30s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e664f9e8aa9846239b250314b7f3d494 : SUCCESS in 49m 23s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/de92ff6f965c4810b67a95811b6e7349 : SUCCESS in 48m 43s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/cc739c404807466693462db9d4dcf331 : SUCCESS in 53m 16s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e593a37675c5405883667cd149f6c906 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d19f223686f4d8f5fd7309452827752304d340d5","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-18 13:09:15.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4cbae42c85ce7fbe1a24f5736cb265ccd47a641d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 13:09:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68c8ac702c163c956cbdcd89ed11f11173c0f19f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 14:02:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8fe5e48892ed01eed25c918f25fdbe7ffa7ca6d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 14:02:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfbe621cf0de41efbcf9013926921d48 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd2e11b4805e4ca4b39781aab02ead0e : SUCCESS in 6m 45s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5995fd4949914315975f34cccd0e21d2 : SUCCESS in 45m 37s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/353d2669a32e4e2781090ee9f8c92c55 : SUCCESS in 46m 50s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a8bafa47cebe45a3843dbea7a6e76384 : SUCCESS in 49m 52s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/79ddda66e9f545ef846d6df1c46dae7f : SUCCESS in 37m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08a67720db068f41358ea6e863aa12639dc4a534","revisions":{"08a67720db068f41358ea6e863aa12639dc4a534":{"kind":"REWORK","_number":1,"created":"2020-06-18 11:55:52.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/67/736667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/67/736667/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/67/736667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/67/736667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/67/736667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/67/736667/1"}}},"commit":{"parents":[{"commit":"ad67a6c01b36da0b8d8987b210727f119947686c","subject":"Update rake version in puppet module Gemfile, fix linter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ad67a6c01b36da0b8d8987b210727f119947686c"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-18 11:51:26.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-18 11:55:50.000000000","tz":120},"subject":"Let apache module purge config","message":"Let apache module purge config\n\nOtherwise can cause issues when enabling prefork as mpm_event is enabled\nby default in httpd. A recent change in puppet-apache [1] has added\nhttpd.modules.conf directory to load, which includes default mpm config\nthat conflicts with prefork.\n\nIn the past we disabled purge because of issues with nagios [2], however\nnagios deployment is not longer supported by packstack, so we can safely\nenable apache config purge which is the default option.\n\n[1] https://github.com/puppetlabs/puppetlabs-apache/commit/6425cc46a841fa6ea5bb3f34b8f31d86c14e6914\n[2] https://github.com/redhat-openstack/packstack/commit/686625e17b19c31b9640c64263d9d3bb2bab48c7\n\nChange-Id: Ifdccd8609658b3170b8356bddc5062f252677d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/08a67720db068f41358ea6e863aa12639dc4a534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/08a67720db068f41358ea6e863aa12639dc4a534"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
