)]}'
{"id":"x%2Fpackstack~736715","triplet_id":"x%2Fpackstack~master~I9808793db64f9e4ad6d2af818d52378f04d03456","project":"x/packstack","branch":"master","topic":"retire-neutron-fwaas","hashtags":[],"change_id":"I9808793db64f9e4ad6d2af818d52378f04d03456","subject":"Remove FWaaS deployment","status":"MERGED","created":"2020-06-18 13:33:47.000000000","updated":"2020-06-22 10:46:16.000000000","submitted":"2020-06-22 10:46:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736715-1592822776171-bef71893","meta_rev_id":"4217001ee31504e52da70b1b9709d2d3bf70cbf6","_number":736715,"virtual_id_number":736715,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-22 10:46:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-06-22 09:32:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-18 15:29:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2020-06-22 09:32:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 15:29:36.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-06-22 09:32:42.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-06-22 10:46:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6f10856a065d0be6a4133d9e90598eb9739c461","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-18 13:33:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c5afae0ec0295433e11cb605ea4dfdd37bb3f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 14:21:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a5df52a36064c248aebe591260b5579 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82efef1004a44325ae9f681cc957cc5b : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5933e183b693439589cd1730037206e6 : SUCCESS in 5m 31s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/82baa211ca9b40c9a82ecb14ff058d54 : SUCCESS in 45m 32s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/147a9d1365744bb09857ae9a0bf68103 : SUCCESS in 44m 14s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ae8087f8112e4cd0a39f9ca8d7b80136 : SUCCESS in 31m 55s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bf465d958cff45fbbeb7d85cb43d23de : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c3512721d88cd7cf40dd5e7296d20b7b7bffbbd2","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-18 15:29:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b9c023eba82268badcedbe6c5804ee25dfe37a8e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-22 09:32:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b74ba23a7e1f15a5c08778ead8f28dbb8679553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 09:32:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7fe13925dcadd599f36f94e0bf8eaa789b78d17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 10:46:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd45c25a0e2742238f19b2558b9c19b3 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7588c5b79ae1481d808a586a16c034d3 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfc2f85a5c6447fb89fe28edf7c10cda : SUCCESS in 5m 51s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/98d4021864bb4187a43e27a675cb59c8 : SUCCESS in 1h 11m 10s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/22907c3b308b4f36a55ba4dfd8bdda12 : SUCCESS in 47m 44s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5c2d9721fa3a4c21b2ebd1cdd5585d1b : SUCCESS in 1h 03m 44s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/601459c38bf7464f850ee4306bfe04fc : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b4e6f08adcbe90284e4950f249abf366da45340d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 10:46:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df280619e583c927d4eb814c733fec10da5a1387","revisions":{"df280619e583c927d4eb814c733fec10da5a1387":{"kind":"REWORK","_number":1,"created":"2020-06-18 13:33:47.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/15/736715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/15/736715/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/15/736715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/15/736715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/15/736715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/15/736715/1"}}},"commit":{"parents":[{"commit":"08a67720db068f41358ea6e863aa12639dc4a534","subject":"Let apache module purge config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/08a67720db068f41358ea6e863aa12639dc4a534"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-18 13:31:22.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-18 13:31:22.000000000","tz":120},"subject":"Remove FWaaS deployment","message":"Remove FWaaS deployment\n\nFWaaS is being retired upstream during Victoria release, so we need to\nremove it from packstack options.\n\n[1] https://review.opendev.org/#/c/735829/\n\nChange-Id: I9808793db64f9e4ad6d2af818d52378f04d03456\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/df280619e583c927d4eb814c733fec10da5a1387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/df280619e583c927d4eb814c733fec10da5a1387"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
