)]}'
{"id":"x%2Fpackstack~738223","triplet_id":"x%2Fpackstack~master~I52435059d25d6d02b74186c5215e2659f153e412","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I52435059d25d6d02b74186c5215e2659f153e412","subject":"Remove warning for Network Manager is enabled","status":"ABANDONED","created":"2020-06-26 15:09:59.000000000","updated":"2021-11-22 06:45:31.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8a9cb1f1bb465a54ffdfa495adfaad1ef37e5432","_number":738223,"virtual_id_number":738223,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-29 09:17:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-29 08:12:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":-1,"date":"2020-06-29 12:30:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 15:20:57.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-06-29 09:17:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-29 09:58:12.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"33741d3061e8eca6881370e1f00a2c7f5a763f7c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-26 15:09:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"793b171013b9e2704d51751d64ccaa46e0e5333e","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-26 15:20:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dee6ebcc13caf5b6d67701bb3f44e3865e8aaafa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 15:26:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c30b2e68f00648999448c658ddac16f3 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd662798742b43e08768cff51cfbe5c1 : SUCCESS in 5m 38s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/335a82f935b24349842b446bd38309f2 : POST_FAILURE in 10m 16s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f2359509e9df4a94b65be84b01827c01 : POST_FAILURE in 10m 05s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/05d15fe0db0c4e75bf774f9c442a2df9 : POST_FAILURE in 11m 44s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a2699160cc1f45febc92666bd5fa7eb6 : POST_FAILURE in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9ac2b31af6e1319321d02010bc4edf504c346f16","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-29 04:30:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"565528e6f9593fe372a3742267de838a33fc17d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 04:46:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ae5b3d88b740acbc850448023c6cd0 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b84d0c0c12e41dabed35b277ccbafb5 : SUCCESS in 3m 31s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4ac9650e36b646afa258f7963aa779ea : POST_FAILURE in 9m 24s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/df12cbfdff6e4f0ab9ac4c88d0227fd2 : POST_FAILURE in 4m 24s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/01b12a0f983141d8abcfffad2cf1704a : POST_FAILURE in 8m 53s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5275b231dbbc4aa986e70f33469cb66a : POST_FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"adb284616b3649d59125a07f94b72f7c74e0a470","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-29 08:01:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e71572cf2496af13eabd8d831c2f591facdde41f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-29 08:12:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"131f3a32fcd4ddc5ff92346b918178b40481c454","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-29 08:12:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ddd1351b206008faabcc3d9002650c1f57175595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 09:17:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5e80c3b11849d0b38f80fb18c7525e : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72da5f02ea134cc8bbae65dfe74e001f : SUCCESS in 4m 20s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8c49b3f88aad4f9d9876f42da705c309 : SUCCESS in 45m 29s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/adf6551ee5f94dd7ae048963ce794723 : SUCCESS in 44m 02s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/167c3f415b9d43248f683e6e49695281 : SUCCESS in 57m 36s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3ce996c33d9d4a4097bf1440ed44c991 : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a802edbc50a38a8ca95c3214c5c6cb8fab2ad385","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-29 09:58:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b987145a9d816469283b5aeffce0a143c71a8e4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-29 12:30:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"561ef32ccdbe33c33cdb1cb4b1cfb9e5b86e242d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-29 12:30:30.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"8a9cb1f1bb465a54ffdfa495adfaad1ef37e5432","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-11-22 06:45:31.000000000","message":"Abandoned\n\ncurrently NM switch is not done, this can be restored later when needed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f5a9708c5e96bdfffc49a4316444ebb36cdbdff9","revisions":{"260dfdecb1d400532e4cf23e7deaa7b44e1fa267":{"kind":"REWORK","_number":1,"created":"2020-06-26 15:09:59.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/23/738223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/23/738223/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/23/738223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/23/738223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/23/738223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/23/738223/1"}}},"commit":{"parents":[{"commit":"df280619e583c927d4eb814c733fec10da5a1387","subject":"Remove FWaaS deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/df280619e583c927d4eb814c733fec10da5a1387"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-06-26 15:04:40.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-06-26 15:04:47.000000000","tz":330},"subject":"Remove warning for Network Manager is enabled","message":"Remove warning for Network Manager is enabled\n\nNetwork-script is deprecated in favor of NetworkManager and not\ninstalled by default in CentOS8. Since Ussuri+ releases\nare CentOS8 only, remove the check and warning for NetworkManager.\n\nChange-Id: I52435059d25d6d02b74186c5215e2659f153e412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/260dfdecb1d400532e4cf23e7deaa7b44e1fa267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/260dfdecb1d400532e4cf23e7deaa7b44e1fa267"}]},"branch":"refs/heads/master"},"f5a9708c5e96bdfffc49a4316444ebb36cdbdff9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-29 08:12:35.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/23/738223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/23/738223/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/23/738223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/23/738223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/23/738223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/23/738223/2"}}},"commit":{"parents":[{"commit":"11691274f25310a8f6ccfa479f4a65a0cd314ab5","subject":"Remove usage of cinder::ceilometer class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/11691274f25310a8f6ccfa479f4a65a0cd314ab5"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-06-26 15:04:40.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-06-29 08:12:35.000000000","tz":0},"subject":"Remove warning for Network Manager is enabled","message":"Remove warning for Network Manager is enabled\n\nNetwork-script is deprecated in favor of NetworkManager and not\ninstalled by default in CentOS8. Since Ussuri+ releases\nare CentOS8 only, remove the check and warning for NetworkManager.\n\nChange-Id: I52435059d25d6d02b74186c5215e2659f153e412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f5a9708c5e96bdfffc49a4316444ebb36cdbdff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f5a9708c5e96bdfffc49a4316444ebb36cdbdff9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
