)]}'
{"id":"x%2Fpackstack~738376","triplet_id":"x%2Fpackstack~master~Ie4432221bdb795dd03f1477609b792ae8400a5fc","project":"x/packstack","branch":"master","hashtags":[],"change_id":"Ie4432221bdb795dd03f1477609b792ae8400a5fc","subject":"Remove usage of cinder::ceilometer class","status":"MERGED","created":"2020-06-29 05:42:13.000000000","updated":"2020-06-29 09:36:41.000000000","submitted":"2020-06-29 09:36:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738376-1593423401480-995c8465","meta_rev_id":"5d949f2133e8b3dc2b9a86b8da18c7bfdc863931","_number":738376,"virtual_id_number":738376,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-29 09:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-06-29 08:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-29 08:40:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-29 08:40:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 08:39:02.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-06-29 08:40:35.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-06-29 09:36:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6caab6c95b861a3516cca342b7e85bb0c44dd068","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-29 05:42:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0aadf7716abbc65e895c32d4f29925a92849d0bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 05:57:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e0d31e37bf4183b9830a8104f5e2b2 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8817b078d8ce4bcdb73d30559199ed97 : SUCCESS in 3m 41s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/91218821e9f54cd48100c53db2199d67 : POST_FAILURE in 9m 45s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d38acda0af3c4d7fb0ae9f360f5fc038 : POST_FAILURE in 10m 15s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6a25791fd88048198f2cdad1f3f8f9f2 : POST_FAILURE in 9m 15s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/61f9c15a55034bb4b35dd00cb8872359 : POST_FAILURE in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"65e7eadf1a20738a99471137dfe12a56d49e81b1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-29 07:10:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d0e9ee6409a097a44ba77e64db694e4e0bfa5d7","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-29 07:17:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c47434a5ea4c100b35263681d57c34b3f92f5082","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-29 07:25:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f7a99b1b63758711f4052ee357d7837bd42a955","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-29 07:26:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f6f3a4da6093b685d819ef3c990b6b370cebb11e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 08:30:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70d0fee95923436295dc3aa6aea72005 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca0fab62e7434a088e6acee542f79df2 : SUCCESS in 7m 37s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9de54ac787304cb1aa7f8f65ebf485e3 : SUCCESS in 55m 18s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/11e4508c24da4e5ba1c6e02e9b140188 : SUCCESS in 43m 51s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3867b485c4c947e9bbf487360035ff9b : SUCCESS in 50m 20s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e51df7f3cc3746c19cf605a06468ca18 : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":3},{"id":"af29dc468bcb0ba55d85e2871f7b5a2c3739fd6e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-29 08:39:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8ab402be6fae500c2916db3cb161265b20038d89","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-29 08:40:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a2680e01ad33097c65b5d60bd7f2de78da6d8ef7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 08:40:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f608701db8abd174347f54e1f60413a80fd3efe2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 09:36:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5d949f2133e8b3dc2b9a86b8da18c7bfdc863931","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 09:36:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1070c8066d0f4dd19e82254b327ac14f : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e0f4168208a4063a3a80491607b91b0 : SUCCESS in 5m 18s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/fb58e2f026ab4f559b4fe51a43168f38 : SUCCESS in 50m 29s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/92915a8c45164c09b6df8a8782bd5e7c : SUCCESS in 48m 47s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f50cc2a8cf3049ae9c7ccd3b0058a3c6 : SUCCESS in 52m 21s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/349e1a0cd31b4baf859a75b559024b4e : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"11691274f25310a8f6ccfa479f4a65a0cd314ab5","revisions":{"112a8401af626b4ef21a34e062b095c044fea33f":{"kind":"REWORK","_number":1,"created":"2020-06-29 05:42:13.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/76/738376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/76/738376/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/76/738376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/76/738376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/76/738376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/76/738376/1"}}},"commit":{"parents":[{"commit":"df280619e583c927d4eb814c733fec10da5a1387","subject":"Remove FWaaS deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/df280619e583c927d4eb814c733fec10da5a1387"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-06-29 05:34:08.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-06-29 05:41:52.000000000","tz":330},"subject":"Remove usage of cinder::ceilometer class","message":"Remove usage of cinder::ceilometer class\n\nThe class was deprecated and now removed with [1],\nthis patch cleans it\u0027s usage and use recommended\nclass to set notification driver.\n\n[1] https://review.opendev.org/#/c/737492/\n\nChange-Id: Ie4432221bdb795dd03f1477609b792ae8400a5fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/112a8401af626b4ef21a34e062b095c044fea33f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/112a8401af626b4ef21a34e062b095c044fea33f"}]},"branch":"refs/heads/master"},"870559aef4898d35f615e9e3fc5fc6523677d7fe":{"kind":"REWORK","_number":2,"created":"2020-06-29 07:10:25.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/76/738376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/76/738376/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/76/738376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/76/738376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/76/738376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/76/738376/2"}}},"commit":{"parents":[{"commit":"df280619e583c927d4eb814c733fec10da5a1387","subject":"Remove FWaaS deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/df280619e583c927d4eb814c733fec10da5a1387"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-06-29 05:34:08.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-06-29 07:09:40.000000000","tz":330},"subject":"Remove usage of cinder::ceilometer class","message":"Remove usage of cinder::ceilometer class\n\nThe class was deprecated and now removed with [1],\nthis patch cleans it\u0027s usage and use recommended\nclass to set notification driver.\n\n[1] https://review.opendev.org/#/c/737492/\n\nAlso remove epel repos as packages from it\nconflicts with RDO repos.\n\nChange-Id: Ie4432221bdb795dd03f1477609b792ae8400a5fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/870559aef4898d35f615e9e3fc5fc6523677d7fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/870559aef4898d35f615e9e3fc5fc6523677d7fe"}]},"branch":"refs/heads/master"},"11691274f25310a8f6ccfa479f4a65a0cd314ab5":{"kind":"REWORK","_number":3,"created":"2020-06-29 07:26:34.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/76/738376/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/76/738376/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/76/738376/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/76/738376/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/76/738376/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/76/738376/3"}}},"commit":{"parents":[{"commit":"df280619e583c927d4eb814c733fec10da5a1387","subject":"Remove FWaaS deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/df280619e583c927d4eb814c733fec10da5a1387"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-06-29 05:34:08.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-06-29 07:24:59.000000000","tz":330},"subject":"Remove usage of cinder::ceilometer class","message":"Remove usage of cinder::ceilometer class\n\nThe class was deprecated and now removed with [1],\nthis patch cleans it\u0027s usage and use recommended\nclass to set notification driver.\n\n[1] https://review.opendev.org/#/c/737492/\n\nAlso remove epel repos as packages from it\nconflicts with RDO repos.\n\nChange-Id: Ie4432221bdb795dd03f1477609b792ae8400a5fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/11691274f25310a8f6ccfa479f4a65a0cd314ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/11691274f25310a8f6ccfa479f4a65a0cd314ab5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
