)]}'
{"id":"x%2Fpackstack~738463","triplet_id":"x%2Fpackstack~master~I1510c3d3549d39dd251b123f80ea2ddf863b7fd1","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I1510c3d3549d39dd251b123f80ea2ddf863b7fd1","subject":"[TEST DNM] Testing test-release-openstack","status":"ABANDONED","created":"2020-06-29 15:36:47.000000000","updated":"2020-06-30 09:44:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed20b46c42777973f8f0971c16eed5c05a85fc98","_number":738463,"virtual_id_number":738463,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-29 15:44:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 15:44:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27caf9d56fc0fb90092637643ce2cf728aa50755","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-29 15:36:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66c247d86949132ccd557024336a03eddd9a8d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 15:44:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/85ce73f1a9e74b35ab11a4b50583c86a : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ed20b46c42777973f8f0971c16eed5c05a85fc98","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-30 09:44:43.000000000","message":"Abandoned\n\nMerged into https://review.opendev.org/735137","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5059e8424ec8d2e067ab589604396eb425e17756","revisions":{"5059e8424ec8d2e067ab589604396eb425e17756":{"kind":"REWORK","_number":1,"created":"2020-06-29 15:36:47.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/63/738463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/63/738463/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/63/738463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/63/738463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/63/738463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/63/738463/1"}}},"commit":{"parents":[{"commit":"11691274f25310a8f6ccfa479f4a65a0cd314ab5","subject":"Remove usage of cinder::ceilometer class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/11691274f25310a8f6ccfa479f4a65a0cd314ab5"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-06-29 15:35:35.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-06-29 15:36:45.000000000","tz":120},"subject":"[TEST DNM] Testing test-release-openstack","message":"[TEST DNM] Testing test-release-openstack\n\nThe version of wheel installed in the current Ubuntu Bionic nodes\nused by the test-release-openstack job does not seem to generate\na proper wheel when we use markdown for the long description. Testing\na change to rst.\n\nChange-Id: I1510c3d3549d39dd251b123f80ea2ddf863b7fd1\nSem-Ver: api-break\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5059e8424ec8d2e067ab589604396eb425e17756"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5059e8424ec8d2e067ab589604396eb425e17756"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
