)]}'
{"id":"x%2Fpackstack~738556","triplet_id":"x%2Fpackstack~master~Iaba8ff454e01cd5c7455da5d0548d8cfeea3f8d8","project":"x/packstack","branch":"master","hashtags":[],"change_id":"Iaba8ff454e01cd5c7455da5d0548d8cfeea3f8d8","subject":"Replace nova_metadata_ip by nova_metadata_host","status":"MERGED","created":"2020-06-30 07:58:36.000000000","updated":"2020-06-30 11:23:44.000000000","submitted":"2020-06-30 11:23:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738556-1593516224040-b4231495","meta_rev_id":"1973fc0e15fd002427bc2d007575fc923e55ca08","_number":738556,"virtual_id_number":738556,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 11:23:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-06-30 10:21:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-30 09:45:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2020-06-30 10:21:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 09:45:45.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-06-30 10:21:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-06-30 11:23:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"784a6ef43da4fa6f259c515322886b6f5ccd8779","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-30 07:58:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b0127269ba17de903acaa85ff8f72b2ba546d09","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-30 08:23:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9810b8078ba6d1bbb1e7fc13455d8e863a04de8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 09:27:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e5b395733a542beb65309e693d2f3a6 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/939e2ed8edf9499bb604b2ff2c3349b0 : SUCCESS in 3m 48s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b19bffa7d0bc49eaaf4d7816b318d1bf : SUCCESS in 52m 22s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/98de209d04e149b3b83c33697657db03 : SUCCESS in 52m 48s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f91c99f6552b4247a39fe0a1552fc835 : SUCCESS in 51m 32s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0d2d57ce5c4a41b48f24655885eb57b6 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9df5baa4ec68b19a28657e30574df2cf3d45301b","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-30 09:45:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dae46cc01c6c6b59470501416ed67d8a305b6880","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-30 10:21:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e9354a7a77eb395d00b303471b05d7470caa0447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:21:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc888b9a0d8d48302916b15fd9bb34fbb2fc6f7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 11:23:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1fc47e3a4954311a55dcc49628f7fb9 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd36a28685484771b6e349f300fa4dd9 : SUCCESS in 3m 48s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c07d9995c82947af89513cb75fc09146 : SUCCESS in 55m 17s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/01addaac9c3e4b4ea9c84774fce1c0a5 : SUCCESS in 51m 06s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/694ae05dec784cb7b3a8176d72618dee : SUCCESS in 52m 00s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c4f9d0a4220b44c0b30431f5878725dc : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":2},{"id":"1973fc0e15fd002427bc2d007575fc923e55ca08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 11:23:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"62a9487c2d2000ef049ac368fd9828bfa706efcb","revisions":{"e03068c3ef688913819064d8de701f39fcebb2f9":{"kind":"REWORK","_number":1,"created":"2020-06-30 07:58:36.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/56/738556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/56/738556/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/56/738556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/56/738556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/56/738556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/56/738556/1"}}},"commit":{"parents":[{"commit":"df280619e583c927d4eb814c733fec10da5a1387","subject":"Remove FWaaS deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/df280619e583c927d4eb814c733fec10da5a1387"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-30 07:56:37.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-30 07:58:34.000000000","tz":120},"subject":"Replace nova_metadata_ip by nova_metadata_host","message":"Replace nova_metadata_ip by nova_metadata_host\n\nnova_metadata_ip was deprecated in favor of nova_metadata_host some time\nago [1] and after moving ovn driver to main neutron repo, it has been\nremoved so we need to use the new parameter.\n\n[1] https://github.com/openstack/networking-ovn/blob/stable/train/networking_ovn/conf/agent/metadata/config.py#L53\n\nChange-Id: Iaba8ff454e01cd5c7455da5d0548d8cfeea3f8d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e03068c3ef688913819064d8de701f39fcebb2f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e03068c3ef688913819064d8de701f39fcebb2f9"}]},"branch":"refs/heads/master"},"62a9487c2d2000ef049ac368fd9828bfa706efcb":{"kind":"REWORK","_number":2,"created":"2020-06-30 08:23:37.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/56/738556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/56/738556/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/56/738556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/56/738556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/56/738556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/56/738556/2"}}},"commit":{"parents":[{"commit":"df280619e583c927d4eb814c733fec10da5a1387","subject":"Remove FWaaS deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/df280619e583c927d4eb814c733fec10da5a1387"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-30 07:56:37.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-30 08:22:24.000000000","tz":120},"subject":"Replace nova_metadata_ip by nova_metadata_host","message":"Replace nova_metadata_ip by nova_metadata_host\n\nnova_metadata_ip was deprecated in favor of nova_metadata_host some time\nago [1] and after moving ovn driver to main neutron repo, it has been\nremoved so we need to use the new parameter.\n\n[1] https://github.com/openstack/networking-ovn/blob/stable/train/networking_ovn/conf/agent/metadata/config.py#L53\n\nChange-Id: Iaba8ff454e01cd5c7455da5d0548d8cfeea3f8d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/62a9487c2d2000ef049ac368fd9828bfa706efcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/62a9487c2d2000ef049ac368fd9828bfa706efcb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
