)]}'
{"id":"x%2Fpackstack~738587","triplet_id":"x%2Fpackstack~stable%2Fussuri~Iaba8ff454e01cd5c7455da5d0548d8cfeea3f8d8","project":"x/packstack","branch":"stable/ussuri","topic":"rdo-ussuri","hashtags":[],"change_id":"Iaba8ff454e01cd5c7455da5d0548d8cfeea3f8d8","subject":"Replace nova_metadata_ip by nova_metadata_host","status":"MERGED","created":"2020-06-30 11:34:07.000000000","updated":"2020-06-30 14:25:12.000000000","submitted":"2020-06-30 14:25:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738587-1593527112935-f3086ce0","meta_rev_id":"e524fe336f44689a2ca2eeb25fdcf2dd37790ecd","_number":738587,"virtual_id_number":738587,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 14:25:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-06-30 13:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-30 13:27:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-30 13:27:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 13:18:34.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-06-30 13:27:09.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-06-30 14:25:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d64187fd84a08fda54285a2d8a742959364b6b84","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-30 11:34:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a37cd7c94f403ca05b35fe8c8287fcdc1b2855f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 11:52:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b8d9bf1f51944fd9dc45e2cf5f0e032 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d091c064ce9743ae90234f00c57fcae3 : SUCCESS in 3m 41s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/74a6b5c28ec54ad59aba695fbf949c96 : POST_FAILURE in 9m 44s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/368f770fedeb431498db647e71f4d8d0 : POST_FAILURE in 11m 46s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/214bf42c2128477cae81e050396f7289 : POST_FAILURE in 10m 06s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d0eac8394c094f4880d01230b92c20ca : POST_FAILURE in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2307aa57cfec2475d060469930923b80d0fd7574","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-30 11:59:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"966b96afcc48bd653034c7a4413c61af85aa28bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 13:00:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9653ac515cd44106a11244afadfa8915 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31e22026a2874aedb0dc714e6a896c2a : SUCCESS in 4m 16s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9eb28ff0725f41cda7b917180642ac2e : SUCCESS in 49m 31s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2271e6b4149c43159c28cb654b6a8fe3 : SUCCESS in 49m 13s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/31db7bf7c4754b0ba8626f7701a68e2c : SUCCESS in 55m 45s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a7f8940897aa4e0f906ccfddb3459016 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":2},{"id":"5fba8d494ba0d6b9e57a4cedf4c536a92821ef38","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-30 13:18:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1f7f1690d89c5462a23f0ab6a2cea2f520f001b9","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-06-30 13:27:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bb6db0bd0f8f4915a20b75a70d0636ef6777aedd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 13:29:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"730001fd9626dd3d30f1db3b60f0a86b07e9c788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 14:25:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa5e1d7cdabd44a8be0281f7e995cfaa : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0eb10025e71c4869a4b14edd2fa071d0 : SUCCESS in 7m 14s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ea385fc455c44a5b935eb43b93d8d42f : SUCCESS in 45m 05s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/231b2c37633f4b45b2379f6747be4dab : SUCCESS in 30m 54s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/bd551d928d1e42258390809d814755bf : SUCCESS in 47m 39s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/65391baed1cf4ddb813ab449b4001631 : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":2},{"id":"7ce7b4880af5511113feada8219e5539ce308698","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 14:25:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e51db4189f80a7a5efa2ec3b5c1b8ed9dc1c9dc7","revisions":{"6a94899fd1bc462ae1010e38fdf3f6b4cfd3e723":{"kind":"REWORK","_number":1,"created":"2020-06-30 11:34:07.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/87/738587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/87/738587/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/87/738587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/87/738587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/87/738587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/87/738587/1"}}},"commit":{"parents":[{"commit":"88f5683a174e07b8f712f618952e003630b3ae5e","subject":"Fix CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/88f5683a174e07b8f712f618952e003630b3ae5e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-30 07:56:37.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-30 11:33:48.000000000","tz":120},"subject":"Replace nova_metadata_ip by nova_metadata_host","message":"Replace nova_metadata_ip by nova_metadata_host\n\nnova_metadata_ip was deprecated in favor of nova_metadata_host some time\nago [1] and after moving ovn driver to main neutron repo, it has been\nremoved so we need to use the new parameter.\n\n[1] https://github.com/openstack/networking-ovn/blob/stable/train/networking_ovn/conf/agent/metadata/config.py#L53\n\nChange-Id: Iaba8ff454e01cd5c7455da5d0548d8cfeea3f8d8\n(cherry picked from commit 62a9487c2d2000ef049ac368fd9828bfa706efcb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6a94899fd1bc462ae1010e38fdf3f6b4cfd3e723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6a94899fd1bc462ae1010e38fdf3f6b4cfd3e723"}]},"branch":"refs/heads/stable/ussuri"},"e51db4189f80a7a5efa2ec3b5c1b8ed9dc1c9dc7":{"kind":"REWORK","_number":2,"created":"2020-06-30 11:59:30.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/87/738587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/87/738587/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/87/738587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/87/738587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/87/738587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/87/738587/2"}}},"commit":{"parents":[{"commit":"88f5683a174e07b8f712f618952e003630b3ae5e","subject":"Fix CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/88f5683a174e07b8f712f618952e003630b3ae5e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-30 07:56:37.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-30 11:59:20.000000000","tz":120},"subject":"Replace nova_metadata_ip by nova_metadata_host","message":"Replace nova_metadata_ip by nova_metadata_host\n\nnova_metadata_ip was deprecated in favor of nova_metadata_host some time\nago [1] and after moving ovn driver to main neutron repo, it has been\nremoved so we need to use the new parameter.\n\n[1] https://github.com/openstack/networking-ovn/blob/stable/train/networking_ovn/conf/agent/metadata/config.py#L53\n\nChange-Id: Iaba8ff454e01cd5c7455da5d0548d8cfeea3f8d8\n(cherry picked from commit 62a9487c2d2000ef049ac368fd9828bfa706efcb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e51db4189f80a7a5efa2ec3b5c1b8ed9dc1c9dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e51db4189f80a7a5efa2ec3b5c1b8ed9dc1c9dc7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
