)]}'
{"id":"x%2Fpackstack~741372","triplet_id":"x%2Fpackstack~master~I68642abc05b74e11e0c264ed5c89b9175e520327","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I68642abc05b74e11e0c264ed5c89b9175e520327","subject":"Remove deprecated params from neutron::notifications","status":"MERGED","created":"2020-07-16 04:52:23.000000000","updated":"2020-07-16 09:30:21.000000000","submitted":"2020-07-16 09:30:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741372-1594891821520-41dc37f0","meta_rev_id":"198d0f6e302ef2c1d2d5f5a0b34587eecae01c32","_number":741372,"virtual_id_number":741372,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-16 09:30:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2020-07-16 08:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2020-07-16 08:25:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-16 06:49:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 06:49:16.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-07-16 08:25:48.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-07-16 09:30:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"578ecd631c502068669c20d872131bb44c3545f4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-16 04:52:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf1e15256e6676392df2997126ea2b8c611bd396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 06:00:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1132e1ffe2545ebb9bf179d834fe3cc : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd788bad0d0e4c9f8d762f93fbb63a7f : SUCCESS in 3m 38s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/994a28d7d2d8461ebf156b22d28d6afb : SUCCESS in 31m 01s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/fc19824d7fe644c8b18108b74f117b0f : SUCCESS in 29m 19s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ce8f40b06a884ddcac84476882f78c1b : SUCCESS in 52m 07s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c72d242533ef44e5bc5024a00955dbc6 : SUCCESS in 50m 44s","accounts_in_message":[],"_revision_number":1},{"id":"39de30464135e488732cd59ddb1106341d06f849","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-07-16 08:25:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c140b20ec78706685dcad8001daa49c032b48b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 08:25:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0be46abae5de0f4371b16ed52a96341457698956","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 09:30:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"198d0f6e302ef2c1d2d5f5a0b34587eecae01c32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 09:30:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4643236b21e24c90b6b0b9690342ac5c : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d5a9f5ac0d74f65b73344c7478ae983 : SUCCESS in 5m 15s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d0fd8185a43a47a3b1a1d403590e4a0d : SUCCESS in 53m 40s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ef160095217f49118dce730db214fb02 : SUCCESS in 56m 32s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8f972be0e411468ca934bb8fb4800011 : SUCCESS in 56m 48s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/fd00a8eb8af74ec3af0ff1df5db5f206 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3da06bd5e26c74ec0962c89e38cffb6550e8260","revisions":{"f3da06bd5e26c74ec0962c89e38cffb6550e8260":{"kind":"REWORK","_number":1,"created":"2020-07-16 04:52:23.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/72/741372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/72/741372/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/72/741372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/72/741372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/72/741372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/72/741372/1"}}},"commit":{"parents":[{"commit":"116c22e0ab5e8c37c551054196fe623d284abdd1","subject":"Remove options to enable GlusterFS volume in Cinder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/116c22e0ab5e8c37c551054196fe623d284abdd1"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-16 04:50:55.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-16 04:50:57.000000000","tz":330},"subject":"Remove deprecated params from neutron::notifications","message":"Remove deprecated params from neutron::notifications\n\nhttps://review.opendev.org/#/c/740611/ removed deprecated\nparameters, use project_name instead of tenant_name.\n\nChange-Id: I68642abc05b74e11e0c264ed5c89b9175e520327\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f3da06bd5e26c74ec0962c89e38cffb6550e8260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f3da06bd5e26c74ec0962c89e38cffb6550e8260"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
