)]}'
{"id":"x%2Fpackstack~741581","triplet_id":"x%2Fpackstack~master~I70e84890246a7ff8bb29218c0628480b5e91a024","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I70e84890246a7ff8bb29218c0628480b5e91a024","subject":"Use database_connection from neutron::db class","status":"MERGED","created":"2020-07-17 06:31:41.000000000","updated":"2020-07-17 09:13:43.000000000","submitted":"2020-07-17 09:13:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741581-1594977223674-dfe663bb","meta_rev_id":"4a7b6f2cbb25ecc9673c24361511c17ba3c2afdb","_number":741581,"virtual_id_number":741581,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-17 09:13:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2020-07-17 08:10:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2020-07-17 08:10:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-17 08:10:48.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-07-17 09:13:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af655e7a3655aa8276607487b68eb64944f3074a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-17 06:31:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a3070dadf336c4205828429a09addac53fa438c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 07:42:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00f15658cc5b4f1a9ec26ef0424e6456 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eacae2b3a614c5db2c4b9a29dc43346 : SUCCESS in 4m 20s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a4e7dd0ea6cc44dcbb3634aa112fbb50 : SUCCESS in 34m 07s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/06266a8c6c0d4540878ae0e194f3a630 : SUCCESS in 35m 29s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8fb5736b56974b55af03274f400aacd3 : SUCCESS in 50m 51s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/aae6e877ed034f4fbdf8d97a74869884 : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e86871d18962afd8ad3dd8244772a7ed8dc27fb5","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-07-17 08:10:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"402943627ebd2ec84271ddc2938695457ba74b60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 08:12:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04d2827eeaa22e76ce97864d781f905c14dcaa98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 09:13:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4a7b6f2cbb25ecc9673c24361511c17ba3c2afdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 09:13:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9544758b5c2c4ff9b148f1d0d802483e : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13c4be0b5372409db274547fd79091e6 : SUCCESS in 4m 19s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/61b92f104d4545479419b11ebde31143 : SUCCESS in 51m 31s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cc17f88c623c45599b51377c6b037331 : SUCCESS in 49m 03s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/45cb66a5be63423594238a24adf1e198 : SUCCESS in 55m 56s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e906919bdac4424697235a161e19abd9 : SUCCESS in 55m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f3a71ff4f8c4dc47a3140a3e1558941558100c1","revisions":{"2f3a71ff4f8c4dc47a3140a3e1558941558100c1":{"kind":"REWORK","_number":1,"created":"2020-07-17 06:31:41.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/81/741581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/81/741581/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/81/741581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/81/741581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/81/741581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/81/741581/1"}}},"commit":{"parents":[{"commit":"f3da06bd5e26c74ec0962c89e38cffb6550e8260","subject":"Remove deprecated params from neutron::notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f3da06bd5e26c74ec0962c89e38cffb6550e8260"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-17 06:30:42.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-17 06:30:53.000000000","tz":330},"subject":"Use database_connection from neutron::db class","message":"Use database_connection from neutron::db class\n\nDeprecated neutron::server::database_connection is\nremoved with [1], let\u0027s use recommended class for it.\n\n[1] https://review.opendev.org/#/c/737497\n\nChange-Id: I70e84890246a7ff8bb29218c0628480b5e91a024\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2f3a71ff4f8c4dc47a3140a3e1558941558100c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2f3a71ff4f8c4dc47a3140a3e1558941558100c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
