)]}'
{"id":"x%2Fpackstack~743472","triplet_id":"x%2Fpackstack~master~I79e93759300fb9ce302896ab287ef528a2d746f0","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I79e93759300fb9ce302896ab287ef528a2d746f0","subject":"Remove usage of ceilometer::keystone::authtoken","status":"MERGED","created":"2020-07-28 07:02:46.000000000","updated":"2020-07-28 09:29:49.000000000","submitted":"2020-07-28 09:29:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743472-1595928589392-5080a3ed","meta_rev_id":"f41b3c184e2b4f2f19370a06a1ec9cbe4bc547cf","_number":743472,"virtual_id_number":743472,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 09:29:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-07-28 07:07:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-07-28 08:28:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-28 08:30:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-28 08:30:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 07:07:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-07-28 08:28:06.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-07-28 08:30:28.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-07-28 09:29:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20c7cff166f14aac852e53ae7657f4dac7dafbb5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-28 07:02:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1909bcdfc329370d0296b9cbddf907260046fd1f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-28 07:07:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"61ff74e1b24930167d910ea290e085337d819781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 08:22:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f343c1d43b684d5ab3dff0caab5130dc : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7842374f748249f0a45653b7f88a76aa : SUCCESS in 3m 54s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d28b5bd56c33426c8c54f3f2dd4d13d0 : SUCCESS in 48m 24s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2b0f766ddcc848e8a92ce130c8f10706 : SUCCESS in 31m 06s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/74738681e69945fc8c8a02247d05d612 : SUCCESS in 52m 44s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cc32ccc51dca49648e6e2b17d2afb239 : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":1},{"id":"29215c4d1041dcf19b3839dfd12abfe49ef38964","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-07-28 08:28:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"89551788904b1ae78c38b819ad8760351f3d5db0","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-07-28 08:30:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2be4a9c32bc5ed098f118ac1da1c73c1c6658aec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 08:30:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a3808114e6d0d992d7ccdc0b0de887d1fcf5b15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 09:29:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f41b3c184e2b4f2f19370a06a1ec9cbe4bc547cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 09:29:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8bf65a4307844c29f9a05f23c425ee9 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9a0e46bd09844bd8baacb4da3e1edd6 : SUCCESS in 4m 07s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/be1d00de9b6d4ad691938362281a01d1 : SUCCESS in 48m 42s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/56171356645f4a19b35bf70dec77ad33 : SUCCESS in 41m 39s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ff0c912bf46e462d9bc5f924a318e110 : SUCCESS in 51m 10s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3300fe089c0a4406b128ff2b484de127 : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"316862acdf149f5660b3003be5a5d75288f8cc67","revisions":{"316862acdf149f5660b3003be5a5d75288f8cc67":{"kind":"REWORK","_number":1,"created":"2020-07-28 07:02:46.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/72/743472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/72/743472/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/72/743472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/72/743472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/72/743472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/72/743472/1"}}},"commit":{"parents":[{"commit":"2f3a71ff4f8c4dc47a3140a3e1558941558100c1","subject":"Use database_connection from neutron::db class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2f3a71ff4f8c4dc47a3140a3e1558941558100c1"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-28 07:01:31.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-28 07:01:34.000000000","tz":330},"subject":"Remove usage of ceilometer::keystone::authtoken","message":"Remove usage of ceilometer::keystone::authtoken\n\nThe class was deprecated as ceilometer no longer\nprovide api and is now removed with [1], cleanup\nit\u0027s usage from packstack too.\n\n[1] https://review.opendev.org/#/c/739304/\n\nChange-Id: I79e93759300fb9ce302896ab287ef528a2d746f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/316862acdf149f5660b3003be5a5d75288f8cc67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/316862acdf149f5660b3003be5a5d75288f8cc67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
