)]}'
{"id":"x%2Fpackstack~745604","triplet_id":"x%2Fpackstack~master~I8e2eb3e0a19aa7968ad5a6a49272e70b47b063b8","project":"x/packstack","branch":"master","topic":"cleanup-deprecations","hashtags":[],"change_id":"I8e2eb3e0a19aa7968ad5a6a49272e70b47b063b8","subject":"Fix glusterfs config option","status":"MERGED","created":"2020-08-10 23:38:56.000000000","updated":"2020-08-11 09:43:51.000000000","submitted":"2020-08-11 09:43:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745604-1597139031975-ab5395dc","meta_rev_id":"e58ccb396fba8a0e075380ec5473288fecdb0f66","_number":745604,"virtual_id_number":745604,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-11 09:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2020-08-11 08:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-11 08:35:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2020-08-11 08:40:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"date":"2020-08-11 08:35:11.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-11 08:35:11.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-08-11 08:35:14.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-08-11 08:40:56.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-08-11 09:43:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ff0523b1fcbc81d583732d18d34e0112a65e17b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-10 23:38:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08dfe1c5603c9972895b8733eeb498ee689e4156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:37:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/731d269fc22c4273b989baf2891ea84e : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecc24f327937477588fbdccb55626b3b : SUCCESS in 3m 01s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e973134a3e0a43cca529cdb6ca8cdeb3 : SUCCESS in 50m 01s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4696e8c770d8484babe920028b3276a4 : SUCCESS in 49m 36s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/99280be4fbf64a99be143afdcab7aba4 : SUCCESS in 53m 14s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/54bccb4b81c74aae82e4a37a52cc1556 : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":1},{"id":"21c0ab8583b499d1b8c5b6b7a56b64f59ae2bd21","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-08-11 08:35:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4bb2bd8268a3bee500f5954e8b4ac05c5a9c9804","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-11 08:40:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e694eef242f49e1c7893644e8c037b28067179ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 08:41:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b49752420b70ac9f7f267b97265d3c89b440417","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 09:43:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c68d0b89b2a4b4a826eccbecf473508 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9769a440b0814b9891e3aa50f8c8c1a5 : SUCCESS in 3m 54s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0b39e66e09b44d54ad85d51aa10c1f89 : SUCCESS in 55m 24s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5f0b260a54e2448587bef4a3a020840e : SUCCESS in 53m 52s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/692836b0a46647928f0e31a28076614f : SUCCESS in 56m 54s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/919f165b4a5b4a35ac3efa80410c9a9e : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0b3adb64cd71f30257a0225c568d8b0dd49cf284","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 09:43:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ddd1f60839c0a1c01abcb55417a1fb6a19fdca6","revisions":{"5ddd1f60839c0a1c01abcb55417a1fb6a19fdca6":{"kind":"REWORK","_number":1,"created":"2020-08-10 23:38:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/04/745604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/04/745604/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/04/745604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/04/745604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/04/745604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/04/745604/1"}}},"commit":{"parents":[{"commit":"0392e3c5636bb048a911d275d380c297343a77d7","subject":"Remove usage of neutron::plugins::ml2::firewall_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0392e3c5636bb048a911d275d380c297343a77d7"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-10 23:38:28.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-10 23:38:54.000000000","tz":-420},"subject":"Fix glusterfs config option","message":"Fix glusterfs config option\n\nThe manila glusterfs option\n\"glusterfs_native_path_to_private_key\"\nis a deprecated name for \"glusterfs_path_to_private_key\" [1]\nand the fallback is being removed in manila. [2]\n\n[1] https://opendev.org/openstack/manila/src/commit/99725abba1a7b0182e36dff761a1920153afbf3f/manila/share/drivers/glusterfs/common.py#L41-L42\n[2] https://review.opendev.org/745206/\n\nChange-Id: I8e2eb3e0a19aa7968ad5a6a49272e70b47b063b8\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5ddd1f60839c0a1c01abcb55417a1fb6a19fdca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5ddd1f60839c0a1c01abcb55417a1fb6a19fdca6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
