)]}'
{"id":"x%2Fpackstack~747836","triplet_id":"x%2Fpackstack~master~I5c0c3804f6a250806bb4f4bfd270e85317c4cdb3","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I5c0c3804f6a250806bb4f4bfd270e85317c4cdb3","subject":"Remove usage of deprecate *_bind_host keystone params","status":"MERGED","created":"2020-08-25 06:28:17.000000000","updated":"2020-08-25 09:01:57.000000000","submitted":"2020-08-25 09:01:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747836-1598346117504-038e877c","meta_rev_id":"ecee5eb959e9d45bd27588a4022da3fe64b2b91e","_number":747836,"virtual_id_number":747836,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 09:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2020-08-25 07:52:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2020-08-25 07:52:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 07:52:50.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-08-25 09:01:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfb609b88f70771c5b4f0f9ec4ee71d469d0c9a4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-25 06:28:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0abf8d992010e81678dd437b200b4a847e2533f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 07:30:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e17d8f0baa5742028781a4ff44a2a68f : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f60f37ab49848b7904c95b007596555 : SUCCESS in 4m 14s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/3020a14315f749e29fff315f39721449 : SUCCESS in 48m 35s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/9553dada47954cb2ae6a164dd4336fc9 : SUCCESS in 50m 59s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1a2b1b4b93b5474fba6e9e0f031dae0b : SUCCESS in 55m 29s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/770122eee46c44888d1cdffee6a6f986 : SUCCESS in 49m 22s","accounts_in_message":[],"_revision_number":1},{"id":"443040d49ae8a60eeb096420b7d75dfb472eea13","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-08-25 07:52:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"42eb00fb283e49b64946daa8865c84da14336bb9","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-08-25 07:52:50.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"275d1e85cbf3d257fbb0e7155a29faad1a88880e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 07:53:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32f17dd6d6c494bf4aa2471346351f3a89671683","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 09:01:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d29a9b33c5f44831a42f83f6baacda11 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7982f25ec002455a876c7ebdec31c998 : SUCCESS in 4m 33s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/efe8a73330624d1d8cc7766afbc1c7ad : SUCCESS in 43m 33s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/427e17268fff423d9d0bb61c60efb19e : SUCCESS in 51m 54s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2513ddadd147422f94e3936b7668a68b : SUCCESS in 1h 03m 51s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/dd22b300fff243639f008cede406f0f9 : SUCCESS in 50m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ecee5eb959e9d45bd27588a4022da3fe64b2b91e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 09:01:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d956ee90c2ab0ea94cd84cf2d9e48406061f885","revisions":{"3d956ee90c2ab0ea94cd84cf2d9e48406061f885":{"kind":"REWORK","_number":1,"created":"2020-08-25 06:28:17.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/36/747836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/36/747836/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/36/747836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/36/747836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/36/747836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/36/747836/1"}}},"commit":{"parents":[{"commit":"5ddd1f60839c0a1c01abcb55417a1fb6a19fdca6","subject":"Fix glusterfs config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5ddd1f60839c0a1c01abcb55417a1fb6a19fdca6"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-25 06:27:10.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-25 06:27:12.000000000","tz":330},"subject":"Remove usage of deprecate *_bind_host keystone params","message":"Remove usage of deprecate *_bind_host keystone params\n\nThe parameters were deprecated earlier and are now\nremoved with https://review.opendev.org/#/c/746648/,\nlet\u0027s cleanup there usage.\n\nChange-Id: I5c0c3804f6a250806bb4f4bfd270e85317c4cdb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3d956ee90c2ab0ea94cd84cf2d9e48406061f885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3d956ee90c2ab0ea94cd84cf2d9e48406061f885"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
