)]}'
{"id":"x%2Fpackstack~748600","triplet_id":"x%2Fpackstack~master~I0ef6b6c96e9604d5a9fb60c360791dda7ca0bf51","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I0ef6b6c96e9604d5a9fb60c360791dda7ca0bf51","subject":"Remove usage of RetryFilter","status":"MERGED","created":"2020-08-28 10:37:01.000000000","updated":"2020-08-28 13:11:49.000000000","submitted":"2020-08-28 13:11:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748600-1598620309173-1545f216","meta_rev_id":"d8bfafcda45526f631106bb7b85540e6fb9c66a5","_number":748600,"virtual_id_number":748600,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 13:11:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2020-08-28 10:57:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2020-08-28 12:03:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-28 12:03:58.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-08-28 13:11:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd70c0cc985c7b15bad3d745612218021b04d50a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-28 10:37:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba9e82940e5b050b9f1794320119d5d10d34e5d0","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-08-28 10:57:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a141cd92cf21385957f3c2e61b27a3638d7a4428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 12:01:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df85220d7947450893b296086998103b : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1d6c401456c4fc4b9fd2ec92eeb0d55 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f123c089822245fd80b71e0b54a612b6 : SUCCESS in 4m 29s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c43bbe2dff2649859fe5deb95b332ba6 : SUCCESS in 47m 32s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d52d63eb2acf4c36ad3843685abddfd1 : SUCCESS in 47m 19s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1d3e727e8bd545f78f68fb9bff3b8279 : SUCCESS in 1h 21m 10s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0b1b9fb178e24a87b70d94b587804bee : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a16939b5b211328fdedca74aa215b3b05ab0da9c","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-08-28 12:03:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f5f3725707051664b0c5be0888b11f7e19e44893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 12:04:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"edd7b840d26df5c9691ffa9f660193b3fa290c9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 13:11:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52cf0c76a71e4f0aa2a9d63f9c35d4bb : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8376e2d67dbe4735aaa7ee81961eace9 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51a538a03ed6428cb04889ab355fe1fc : SUCCESS in 5m 17s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/624b76c729b74895ac69541f32c149cb : SUCCESS in 30m 47s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c648cf66079d4059a8c6f15005589eca : SUCCESS in 50m 17s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/54271062ba2048da85017fe59faa57f0 : SUCCESS in 1h 00m 06s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/10ae2ad9e51e4e698e9392bb53b141bc : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d8bfafcda45526f631106bb7b85540e6fb9c66a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 13:11:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fd13900a23166f49122d1aec63685e6b3e9194a","revisions":{"1fd13900a23166f49122d1aec63685e6b3e9194a":{"kind":"REWORK","_number":1,"created":"2020-08-28 10:37:01.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/00/748600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/00/748600/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/00/748600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/00/748600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/00/748600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/00/748600/1"}}},"commit":{"parents":[{"commit":"3d956ee90c2ab0ea94cd84cf2d9e48406061f885","subject":"Remove usage of deprecate *_bind_host keystone params","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3d956ee90c2ab0ea94cd84cf2d9e48406061f885"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-28 10:35:32.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-28 10:36:01.000000000","tz":330},"subject":"Remove usage of RetryFilter","message":"Remove usage of RetryFilter\n\nRetryFilter was deprecated in Train and is now\nremoved[1] from nova, so let\u0027s cleanup it\u0027s usage.\n\n[1] https://review.opendev.org/#/c/744800\n\nChange-Id: I0ef6b6c96e9604d5a9fb60c360791dda7ca0bf51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1fd13900a23166f49122d1aec63685e6b3e9194a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1fd13900a23166f49122d1aec63685e6b3e9194a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
