)]}'
{"id":"x%2Fpackstack~749044","triplet_id":"x%2Fpackstack~stable%2Fussuri~I76519ef0fe6b5c350b6aec92b204f41e363cf65a","project":"x/packstack","branch":"stable/ussuri","hashtags":[],"change_id":"I76519ef0fe6b5c350b6aec92b204f41e363cf65a","subject":"Replace markdown by rst format","status":"MERGED","created":"2020-08-31 12:14:32.000000000","updated":"2020-09-02 06:27:19.000000000","submitted":"2020-09-02 06:27:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749044-1599028039429-1d2acb16","meta_rev_id":"2ea8d16c0a1c6b4f3e17289cbfbbb9a3c1a5eaa4","_number":749044,"virtual_id_number":749044,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-02 06:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-01 08:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"date":"2020-09-02 04:22:25.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-01 12:37:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 12:37:37.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-09-02 04:22:25.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-09-02 06:27:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"925e6f2bf988b438a4686b96cbf2247bba8815db","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-08-31 12:14:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9de4902170ec83a85026e55894f9653195293a4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 12:52:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0c52f361e15d4a05a55a157a780a74f9 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca7b9c82ea44f0b820bf95bd627dda4 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/869ed3f0d0584ff4856a14fb8eec3f6f : SUCCESS in 3m 19s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b86d98809c8d40dc83d634d88019e221 : FAILURE in 21m 20s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0bfd92c5e5b44bdfade0736781bd4c74 : FAILURE in 20m 17s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/20ff12b66eb7419582000765981c8bdf : FAILURE in 21m 33s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f825a00174cd487db263795c970c29d7 : FAILURE in 32m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2a8f282c8596eac4d7f52afe3f9f55fe259ec7df","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-08-31 13:55:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2d1d8c7a32e2033dcf1b1053b497c78a07082a45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:19:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5f831f071d6240339aed0093f6f9e16a : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f2d8b9430842ca8c38372e490c7aaa : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eb2e2f91ccc4dfbbd20475f9001227f : SUCCESS in 4m 05s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ec7a0610166d48609ed3c35ab101622a : SUCCESS in 51m 42s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cd1aa2cb2b064b8c929400718d5b1481 : SUCCESS in 53m 06s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/52d26a588e48457d8db798d537260bf1 : SUCCESS in 47m 55s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/874041bb3038491da48302c3b587a64b : SUCCESS in 35m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b7c2f34be62c29a587c8f0b4244593a0052fffef","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-09-01 08:33:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2a4c8bac363a1c41cc7eadccd64af24226d8d76a","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-09-01 12:37:37.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"309f503386dfa568ad5a4aebc1e11863b02a3322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 12:38:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"faceed0d3c76cc070d80ae5823e07ecb0e0e18a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:00:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f7ac074be70e4806b712773807dae09f : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f496e28fc2854a9782625efc573ba78d : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4199ae3a4c9e44958c2139e50d22fbce : SUCCESS in 4m 17s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7ceb26081a0b41d6b720c94b7eb21ff6 : SUCCESS in 1h 19m 09s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/03b77940f7fb49cab2869e29861c1120 : SUCCESS in 49m 28s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a112fa5f127f4e42b757474c6ee3996c : SUCCESS in 48m 10s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bedf46b606824a44ac21e67cbfced341 : FAILURE in 16m 59s","accounts_in_message":[],"_revision_number":2},{"id":"366257dd0e671e70b7e8d6d442a4431312231a05","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-09-01 14:42:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"08024fcc54cfa41437a36c10ceff13d5f239d4b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 15:55:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f6c76124b12c4da2b949923545b33737 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d6bc685b9c4060956a7a9bf0a29e32 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a1debe46e7d405c9b5cf914e9269c2e : SUCCESS in 4m 20s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c760098b2ca243d1b97ac288161574b7 : SUCCESS in 1h 05m 14s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7061582848b64046b3cbe8ee3b8d7963 : SUCCESS in 43m 56s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7a6b2a29c20d4d839bf79c946cb1e5b1 : SUCCESS in 54m 11s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cb2468c24b0e419b84e2dcfde566b99e : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":2},{"id":"aa967bf20c1aa07b8e6f7f5760874a02509d2a8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 15:56:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c9fa030592cd4db5ddd6bc71b15d5d27583f02d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 17:04:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3515825f6ade4c0ab6fc49e5e3769c06 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48cdebe1e57d416a88d3fe55796dac8e : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7daf18185aa4348b2a506769468a8fd : SUCCESS in 6m 41s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ad5867c7cca44a3c85de8c837e6e4236 : SUCCESS in 32m 45s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b3063dd75f4c44009ea5ac51aea00e04 : SUCCESS in 49m 29s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/871a5e1bec07420b9017ce962689ec96 : FAILURE in 47m 16s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6338c1fcfc6647c6a1fd2f1f1dd82615 : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4a08d15c18bb2127d17e79030c2415f074a5f5d1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-09-02 04:22:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7efa513782aafaaa8e03b4f4f68a3dc1aaaec576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 05:21:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6674979bf6054fa0bc540105ed493c19 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ddf81ddc85d475baaf940ce95bd10bb : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ab3f7fe25e2414ead23c2b480c5612f : SUCCESS in 3m 37s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b2665bd7c27e4986861daa6f35175bf2 : SUCCESS in 43m 38s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3a3080b906fc466195e7cc31264dd23b : SUCCESS in 45m 40s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/bef4c0ea7d9a48feaa176dfe89592a03 : SUCCESS in 50m 08s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b9c47644d78f45439f0d5f91f723321f : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":2},{"id":"7bd0f437d8eb5eb25378e05fc4427ce75a59251b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 05:21:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9f01ddd969bd1e2f1fed3e6ffe123f3d3bf556d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 06:27:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a6cacfa83d24280894bb992610322a5 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cfe89b978c34ae1bdf291ce7c92666c : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/285f257a3187424288952ff366a399fd : SUCCESS in 4m 59s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a31adf0f67f84b3e81062de926572d27 : SUCCESS in 50m 44s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6031a0c62b1d487a84454a4157cc9aeb : SUCCESS in 50m 36s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/fbc89a67a36c48758560eee5d430736c : SUCCESS in 54m 34s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e272b86ea0d04a30bddfd3771e46b04e : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":2},{"id":"2ea8d16c0a1c6b4f3e17289cbfbbb9a3c1a5eaa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 06:27:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"112924ba26ebbfd44d1bd6cb6f0781f21f46a255","revisions":{"4b35fea8dd19ff787ea6ff91e0fdfa03cf56f642":{"kind":"REWORK","_number":1,"created":"2020-08-31 12:14:32.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/44/749044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/44/749044/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/44/749044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/44/749044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/44/749044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/44/749044/1"}}},"commit":{"parents":[{"commit":"e51db4189f80a7a5efa2ec3b5c1b8ed9dc1c9dc7","subject":"Replace nova_metadata_ip by nova_metadata_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e51db4189f80a7a5efa2ec3b5c1b8ed9dc1c9dc7"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-11 13:41:04.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-08-31 12:14:23.000000000","tz":120},"subject":"Replace markdown by rst format","message":"Replace markdown by rst format\n\nReplacing README.md by rst format as the wheel version in Ubuntu\nBionic nodes seems to not work properly with markdown files.\n\nChange-Id: I76519ef0fe6b5c350b6aec92b204f41e363cf65a\n(cherry picked from commit fad2c393cb7709a38f9389cedcea729720cb841a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4b35fea8dd19ff787ea6ff91e0fdfa03cf56f642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4b35fea8dd19ff787ea6ff91e0fdfa03cf56f642"}]},"branch":"refs/heads/stable/ussuri"},"112924ba26ebbfd44d1bd6cb6f0781f21f46a255":{"kind":"REWORK","_number":2,"created":"2020-08-31 13:55:25.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/44/749044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/44/749044/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/44/749044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/44/749044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/44/749044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/44/749044/2"}}},"commit":{"parents":[{"commit":"e51db4189f80a7a5efa2ec3b5c1b8ed9dc1c9dc7","subject":"Replace nova_metadata_ip by nova_metadata_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e51db4189f80a7a5efa2ec3b5c1b8ed9dc1c9dc7"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-06-11 13:41:04.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-08-31 13:55:07.000000000","tz":120},"subject":"Replace markdown by rst format","message":"Replace markdown by rst format\n\nReplacing README.md by rst format as the wheel version in Ubuntu\nBionic nodes seems to not work properly with markdown files.\n\nChange-Id: I76519ef0fe6b5c350b6aec92b204f41e363cf65a\n(cherry picked from commit fad2c393cb7709a38f9389cedcea729720cb841a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/112924ba26ebbfd44d1bd6cb6f0781f21f46a255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/112924ba26ebbfd44d1bd6cb6f0781f21f46a255"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
