)]}'
{"id":"x%2Fpackstack~750614","triplet_id":"x%2Fpackstack~master~I9c29d0495bc4665595ebfa728b227cc654ada6b6","project":"x/packstack","branch":"master","topic":"test-rdo-repo-updates","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-10-25 16:29:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9c29d0495bc4665595ebfa728b227cc654ada6b6","subject":"DNM test with testing repo added","status":"ABANDONED","created":"2020-09-09 10:00:33.000000000","updated":"2024-10-25 16:29:47.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8fa6c23aa22f5e1e21a51499a8f391ddebdc19b6","_number":750614,"virtual_id_number":750614,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-04 11:45:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-25 06:50:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 04:21:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-11-05 05:39:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 11:32:25.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2021-10-25 06:50:35.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02c556214a5da87537350ff88c407eda1e294905","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-09-09 10:00:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4418f568f35d585269e101f2b4a7e4ed15c8d81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 10:53:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b83300db57504c97af4092ae92bdb8be : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bb9f124fd72494ea19a81a392badc31 : SUCCESS in 3m 43s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4dbcf619bff84c4396b19237563b677c : SUCCESS in 43m 01s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1b93b0291da64d38af34a45cc41014d4 : SUCCESS in 45m 43s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ed24f81fdab34cc8b6b96ed1b205a9be : SUCCESS in 46m 19s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d8586fb757194e9f833d272fd5ae3348 : SUCCESS in 46m 43s","accounts_in_message":[],"_revision_number":1},{"id":"102af5fe54b5a924b5177db73d9770d194c03026","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-09-09 11:00:04.000000000","message":"Topic changed from dnm to nfvsig","accounts_in_message":[],"_revision_number":1},{"id":"c088065b32bc3a4da4308ef947ef6d8fe2cd8076","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-09-16 08:01:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d80bef84b6c541fca6adabea62d20df53f4159b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 08:55:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fb9ed33e9354185914815b57578c55c : FAILURE in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaa8c9b146a2421a8560db177908a988 : SUCCESS in 5m 45s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8a55906ee9b148d6a05bd2c9e7d583ae : SUCCESS in 46m 58s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5b0f45cbbbd44879a79fb6fad9803990 : SUCCESS in 45m 52s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/809d63a6d1f84c539ac04ce53e582256 : SUCCESS in 48m 01s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/711dd85b70f8408e973c476f489c3935 : SUCCESS in 38m 32s","accounts_in_message":[],"_revision_number":2},{"id":"0f95528447029e5f1ff03a1785338a8e3f3b4daf","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-09-18 07:22:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2defc85121c9c513a058bfc66a154101270ae362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 08:26:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc3972436f1d41f48d45c356da954c93 : FAILURE in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25789f206b6746189e1b671476c8ae8a : SUCCESS in 5m 04s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/df22e876bf674a1b9101929ff9927afc : SUCCESS in 49m 36s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f200f6696fa74979b0a7c5567255d533 : SUCCESS in 51m 30s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3cf0cbbc869041e0ac29ec4f6dd481e6 : SUCCESS in 51m 54s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/db5a4ea4a1d240608555a13bc1de91ae : SUCCESS in 48m 28s","accounts_in_message":[],"_revision_number":2},{"id":"5b0f96a5b3e8873ee0ee6d6c3d75e46970c04c6c","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-10-22 09:56:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"82ae89339c7d9f76001f679edfd7dd88fde32771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:12:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/671f6b07336647c8a0ca63d1eb08bc91 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db202a31a74141cfb21eb22980f0e7e3 : SUCCESS in 5m 08s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e091e70277ec4d2b96ff4cf436b9c90c : FAILURE in 13m 19s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c9fc7beaa16342df94fd7724de654cd5 : SUCCESS in 52m 16s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/990ca9da034d4bd09d74fbe17a74cd5d : SUCCESS in 49m 31s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bb2185608ee14f2eb20303a6561f357e : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":2},{"id":"0da3fbe5fbe84ea73a57390fbfda58241686ae2a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-10-23 10:08:33.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c6f9d2e6bab75b292ddb0fd1c1e162dec292aff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 11:26:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e307444f7ef849a2995830d80d5ffd7e : RETRY_LIMIT in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f15974ccb6cb49bd837aa04cbe73b04f : SUCCESS in 5m 37s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/40772c0b186c435a8f90728f5699b39b : SUCCESS in 49m 39s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/563d40af4186405fadf78fcbbc4abc10 : SUCCESS in 52m 04s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e894579d06d640aeb36cbaf3e2a8340f : SUCCESS in 54m 54s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/233687913a53476f8df316c34786bc26 : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":3},{"id":"8d2f6fb431ab16bac5f7001ce3a815068ac5847d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-04 14:28:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e337f8aed1a6d4e77f58337fb423faf2b1d4df24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 14:59:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/055f3534c77845ed8d62986541529060 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87f6fac3e6004adbbe6efe2d758bb4cc : SUCCESS in 3m 21s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/cc779f6e9a7e4084993017b148a4a2ee : POST_FAILURE in 7m 51s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2cb6bd7caa014f1fa2402c26f38a3c57 : POST_FAILURE in 14m 44s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1059a80129af4636bcd4a15edbde2508 : POST_FAILURE in 13m 17s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ecfcf5cda79a44dbb92874c9a4fe8be0 : POST_FAILURE in 15m 02s","accounts_in_message":[],"_revision_number":4},{"id":"d836452da270ad7c161a13ba7e6ca84bdb10f6d6","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-05 04:21:19.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"52274272052b904c0a9bf32437f310373a180a5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 04:42:44.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61889e4663b943c3b1b61092916de6aa : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e61ab22fb81b473b8e937c02bfbbe250 : SUCCESS in 4m 00s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/06c2a9a0df7c4ea8a5c77896778fb71b : POST_FAILURE in 15m 21s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/034a7b98b0234fa6b215bbaf8972e410 : POST_FAILURE in 14m 23s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/61a5d57d0a4e4013b1b3c3ef4df630a5 : POST_FAILURE in 12m 57s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/81747f406f3546ea916cb6c7834932d3 : POST_FAILURE in 14m 48s","accounts_in_message":[],"_revision_number":4},{"id":"7fd934a3a56cb188b1e7325a2b8aabceb613224c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-05 04:49:45.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"caabec097491d78e9377d8f9de94c9ecaf5f2f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 05:39:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c710d9c7807b476385e5ab9f1c1ddd2a : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/065119233b304847bdbdff767d66636a : SUCCESS in 3m 00s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7b35c5a832f24778905fe6639ad4270d : SUCCESS in 41m 41s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/af18e2806e544536838621c812a444f6 : SUCCESS in 44m 26s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c680dfd81cd24f6281a9799c64d4a33c : SUCCESS in 46m 42s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e9cb22ff37f5482e81b3bfda3188cd8e : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":5},{"id":"b5c99c1a2015f467e2665cce8b8dd88349b4fde0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-04 12:25:38.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"29eedab9e9546e48b00359421a9c17801e1c56e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 13:16:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c54f676adab34009a95e53802fd0d4d7 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/683ad547d2c64c0f9bd7742ba1f01541 : SUCCESS in 5m 20s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2ace2656e16f4beebc4ef7d152cc7503 : FAILURE in 31m 50s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0e555e3a37c549adb7265f035f8aa659 : FAILURE in 34m 17s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ab317f9a96c445d6aa8da608704c41fd : FAILURE in 42m 17s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c18c9295cb5f4f20883f7d9e0021460d : FAILURE in 41m 35s","accounts_in_message":[],"_revision_number":6},{"id":"57f239e86076b4a7ae35591c5cc5a13e4b7b7f1e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-12-08 11:32:25.000000000","message":"Patch Set 6:\n\nrecheck .. we are hitting \u0027Error: Unable to find a match: libyaml-devel\u0027 in https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_d8e/765935/2/check/python-tempestconf-tempest-packstack-demo/d8e3276/job-output.txt .. let\u0027s check if the other scenarios have the same problem .. it looks like an issue with PowerTools repo as the package is part of that repo plus I can see in the log ^^ that it does a weird obsoleting step: \n\nObsoleting       : centos-repos-8.2-2.2004.0.2.el8.x86_64             378/379\nwarning: /etc/yum.repos.d/CentOS-PowerTools.repo saved as /etc/yum.repos.d/CentOS-PowerTools.repo.rpmsave","accounts_in_message":[],"_revision_number":6},{"id":"0b6283dfc0a22d06ee03e2859aa4a243c6fac9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 11:57:50.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d14c735880264991a685432227de8fdf : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83103b2f47ea495d851a539384cbf9eb : SUCCESS in 3m 36s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/11b97416adc64e73a98bde5da0a66d44 : POST_FAILURE in 14m 10s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d34bb812a8244285a92fa669e8b4fba9 : POST_FAILURE in 16m 25s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/bfb5953df1b44b758b5d98e1e85b3872 : POST_FAILURE in 14m 01s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6f92d945de6e432383e90c81ab0142c4 : POST_FAILURE in 18m 21s","accounts_in_message":[],"_revision_number":6},{"id":"eb32b6befa90334cf17acf29dcf61a9a24009bed","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-11 06:28:10.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6:\n\u003e \n\u003e recheck .. we are hitting \u0027Error: Unable to find a match: libyaml-devel\u0027 in https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_d8e/765935/2/check/python-tempestconf-tempest-packstack-demo/d8e3276/job-output.txt .. let\u0027s check if the other scenarios have the same problem .. it looks like an issue with PowerTools repo as the package is part of that repo plus I can see in the log ^^ that it does a weird obsoleting step: \n\u003e \n\u003e Obsoleting       : centos-repos-8.2-2.2004.0.2.el8.x86_64             378/379\n\u003e warning: /etc/yum.repos.d/CentOS-PowerTools.repo saved as /etc/yum.repos.d/CentOS-PowerTools.repo.rpmsave\n\nohhk 8.3 release messed it, with release package updated enabled repos(PowerTools) is getting disabled and causing issues.","accounts_in_message":[],"_revision_number":6},{"id":"178b6019712f9f720fec6a0e2060565e408387ce","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-11 06:28:48.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"bc7c24e751096ed732d672a394dac186117375b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 07:01:15.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c043dcb886c4692b4e2633667d6a900 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9532abfe60e545a5a2a040b1711f49b8 : SUCCESS in 4m 57s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4be8f41362c44f8788c3e246628b2c89 : POST_FAILURE in 13m 53s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bedcdf413e37469d980f10fdc09fede3 : POST_FAILURE in 9m 25s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/87f37b0efd6c4b2bac56cc571812b9bd : POST_FAILURE in 25m 35s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9f2f14e5c9a9492886d3332c3786b6aa : POST_FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":6},{"id":"471bc083750e819ca4ab0e90c6f180e952c09938","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-11 10:59:02.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"8995b3f4ccad04f1856c29a7950c2d436fc3c4e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 11:55:48.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/139ab934de18425b91d366f924591430 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1735cd2961684557b291d7b94091e952 : SUCCESS in 3m 32s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1bc8b542d76e40218628dc4ef4412978 : FAILURE in 32m 40s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5938b937cf564f30a36859f3408f0a90 : FAILURE in 33m 10s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a5b3aaf2ebb14fe8847eb83d33f92fec : FAILURE in 36m 16s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/04e14ecfb6e54463a15158c065341bf4 : FAILURE in 53m 02s","accounts_in_message":[],"_revision_number":6},{"id":"44698f3e1c1a10eec9820015710bdb447578189a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-11 11:56:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"82574e04a8d0f1bac09f9b63b2a4580698bde2af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 13:05:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d55b77e11894886b674e0c57fe4274b : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ee9fe46ee6d4f1c85bc2e2cdd563742 : SUCCESS in 4m 33s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/833c5aa7688148a7a1a4a3e0c623cdb7 : FAILURE in 42m 17s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/87dc367747774539859603a75a428bb4 : FAILURE in 45m 54s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2144c88960b04b56af406867d4dbe17a : FAILURE in 48m 42s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/62cb932625a445c4af545dfe45681515 : FAILURE in 1h 03m 14s","accounts_in_message":[],"_revision_number":7},{"id":"e110d7329a6d4becfc59c751a42b67434a393417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-15 06:31:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a7025e19674c5f04fa42eccc064c7791608f50e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 07:54:45.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a32eb7e91f849c68f2d7656c486d836 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/808f68b9a6d847bc9d1df5d2cec6622f : SUCCESS in 3m 26s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7e3caf4ddc0245c48a00f272478a26bc : FAILURE in 48m 14s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/03a451637f3349e4bffffdac36fb8ad8 : SUCCESS in 57m 01s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/95758a68a00546c4b28127ab241d691e : SUCCESS in 1h 06m 24s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/60415ef6db2f4469a707ba4a3dc5d79e : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":8},{"id":"d8e2a6aa233d55f9e5f3b773a3d5cb23854239b7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-16 07:17:13.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"3e7f8edc6a7ae3d13bcd64eb2ce2f07c7c83164d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 10:42:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd22d8d0d3ae45e7b7287565b0a3caf3 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b0ecf098c2f41a49536547a6e727758 : SUCCESS in 4m 46s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/37eb90bb2d4c43949af8d413e12ce512 : SUCCESS in 57m 32s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2cdd94e183db4581a1c0c79d44362251 : SUCCESS in 54m 37s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6b162d134dcc4e18a95a433c0c32f15a : SUCCESS in 59m 26s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9e917b447c78409b943d18987ffad677 : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":8},{"id":"de2c484a460ad03187931ab35542523c794708f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-02-05 07:11:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f1151c980df2a8a9ff4d7701e5452bdea51dbaa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 08:09:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f67fa9d407144cf9b01e6e5b0effdfc4 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/119a1c8d016842f4b69b7c7d1c65d7f7 : SUCCESS in 3m 18s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7b03297466754395be1e078846b0e480 : SUCCESS in 46m 18s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5488edfe332d40b39eba47848c3da8d0 : SUCCESS in 50m 17s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a771b04b1aee43dea79cc49b7e3b9a62 : SUCCESS in 48m 16s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4373916e14c845b3ab4c95a95e3547dd : SUCCESS in 48m 50s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1f23aa20497f407dbc6bb36955eb371c : SUCCESS in 47m 59s (non-voting)\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d2a6ac1f16a343bda1b463658d49404e : SUCCESS in 47m 52s (non-voting)\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/07e57a85114b43a99834b7ef619d0646 : SUCCESS in 50m 56s (non-voting)\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0216752eab8945a98ca08e7c91e5ba72 : SUCCESS in 51m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"44f77412d193503fcc686b2f719cd0ebf2649b1c","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-03-05 10:15:14.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"79e55cb908e69d8ed79191e6382effd2a9b17710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 11:23:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeff1ccbefd04394b9f262eb32b1edc7 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b769fdf8e2ed4974a8f5ee9367ab70aa : SUCCESS in 4m 37s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9d3f2fd1869c40b7933ce96b76364553 : SUCCESS in 39m 34s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/029a77d2acd742e68e8b199d93b62049 : SUCCESS in 44m 04s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9dc55fefabd54136bea2e2835a8cfdaa : SUCCESS in 1h 02m 36s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f4c3cd342d3e4c59970051687b53f3ea : SUCCESS in 58m 47s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a0eddc82e354460cb2ce802a76409cae : SUCCESS in 39m 06s (non-voting)\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/826876a1d8024eca8ae6ac81318902fa : SUCCESS in 46m 59s (non-voting)\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5431867dbc574df1baaeb94ad3f5ff75 : SUCCESS in 43m 42s (non-voting)\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cebf971643714a2cab64838574fe7957 : SUCCESS in 54m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"620cea1d49f0370037ba73ea8892a9d80cf6441e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-04-12 13:26:37.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"485099195828678d7524964ebcee49a0d34c046a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 14:22:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f2cc9e0da064b718e268d33b5ccd0c3 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dab9e55915974b059491baaadd86bdb1 : SUCCESS in 4m 41s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a28352030c794a6988264841cb1b72d1 : SUCCESS in 42m 23s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/56539e29f0bc4186b9a235fdf3135817 : SUCCESS in 47m 48s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a71bbb4fca36426b916bcd98af5727c9 : SUCCESS in 49m 11s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9b971d916ebd4a448a98bf1f0c435260 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":10},{"id":"9f32de6604d2fbeff2e61415f493b72d571fb517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-09-01 09:04:12.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"c968803ef781fd8c86d9a6b63e869dc6fe6e90fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 09:33:38.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d36ea42ae57345bab2408d55f2e30232 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59d672958d874c5b95a6d116b08b8158 : SUCCESS in 5m 54s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5ec58165918e45cd8b8ef140287e8443 : POST_FAILURE in 19m 23s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/31f10bfd307b4506ba777f2ec7de1724 : POST_FAILURE in 17m 37s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/769445fb677c43ae8dcb75615f1e0be6 : POST_FAILURE in 18m 09s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2d845d3cc0fe43da9b5382e6da14ced1 : POST_FAILURE in 19m 23s","accounts_in_message":[],"_revision_number":11},{"id":"8ec4df60a50d627fde30deb142e505e43824cbf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-09-01 12:03:06.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"52d6b190a7d6bc26b33385723023b942bca0c5dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 13:06:37.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/873502e01b754efa859670e5329f30d5 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ef2a56afd4e4af08c27075de98b8142 : SUCCESS in 4m 05s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/29cc41de3d8543f3a7ffa5860f0b0d8b : SUCCESS in 47m 52s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c022c6246ba04169a71f48085e0ceccb : SUCCESS in 50m 54s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/455a2774e749477f82bffb77110fa70d : SUCCESS in 51m 18s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9857c7143dec4d4f8b29df99810c2015 : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":12},{"id":"0066b1d0760b1d32b4aa98c0680414a9bcd410e1","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-09-01 16:05:18.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"11068ec80db8cd77ae76d748cac8740116375f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 17:12:47.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c1a314aea944a3a9ebf7afcff0d3fd7 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0ce7478f18146efac5e731718206f71 : SUCCESS in 6m 27s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5811d6527a9d499fb4f4720fdfd330a0 : SUCCESS in 46m 28s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2a04bc5e133a42d19adf0b2d69a9c091 : SUCCESS in 50m 51s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5d8ad71bd3924df5833405ee5e6e3310 : SUCCESS in 51m 03s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7a3444845ebf4d93a716b99bc3c0ec4d : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":12},{"id":"45f274811a8db0d4b685da42cc3489f61dbaab1b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-10-25 06:37:34.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e8512f94c2405415871ab87b674b7d068b219347","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-10-25 06:50:35.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- packstack-centos9-integration-scenario001 https://review.rdoproject.org/zuul/build/45a66627ec9749159707901ecfc2cca4 : RETRY_LIMIT in 2m 27s (non-voting)\n- packstack-centos9-integration-scenario002 https://review.rdoproject.org/zuul/build/9acf12e67f4f4352b8c5f4d31db57b6c : RETRY_LIMIT in 2m 19s (non-voting)\n- packstack-centos9-integration-scenario003 https://review.rdoproject.org/zuul/build/f72b447e3d3a467cadfe2148f84619b3 : RETRY_LIMIT in 2m 26s (non-voting)\n- packstack-centos9-multinode-scenario002-tempest https://review.rdoproject.org/zuul/build/694b152bedbb433eacf92aa174e0472e : RETRY_LIMIT in 2m 48s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a0b5aa15d5f71f8d78c6723f179604f11e05dfec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 07:45:21.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff0c3e9ce6fe4fc1b600d3d67d2a95b5 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa69aff777c74ba998c973a1870bd1fc : SUCCESS in 7m 09s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0bc9f3ee12104122aeb01de26af485d7 : SUCCESS in 38m 34s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/15f9977754f14cf384c81f73cbca89e1 : SUCCESS in 43m 13s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/163dee95cfe342abb7741aa2edcebfe0 : SUCCESS in 1h 02m 29s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7d46928e551d4aabb575535d09fe8e1c : SUCCESS in 52m 33s","accounts_in_message":[],"_revision_number":12},{"id":"4552ca5c7ef4f8ec17f94328446f5faceeb43b94","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-11-04 10:46:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"81012832768a6530be77240deed9e0d5185a2c99","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-11-04 11:02:05.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- packstack-centos9-integration-scenario001 https://review.rdoproject.org/zuul/build/cea1fe9d824147609f4a657b3aaadf72 : RETRY_LIMIT in 2m 15s (non-voting)\n- packstack-centos9-integration-scenario002 https://review.rdoproject.org/zuul/build/4340ba67115948ceab4c92baf660d399 : RETRY_LIMIT in 2m 49s (non-voting)\n- packstack-centos9-integration-scenario003 https://review.rdoproject.org/zuul/build/b9c63c83ffee4530a4614456230fc936 : RETRY_LIMIT in 2m 20s (non-voting)\n- packstack-centos9-multinode-scenario002-tempest https://review.rdoproject.org/zuul/build/68053b4bd3254a0b979b45705c2c4b32 : RETRY_LIMIT in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"7da9da0a7ec77483da912403aa45d6660473e024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 11:45:31.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b7985c82ee41af8a6a00809aa105c4 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a49a116464584c4bbb3ad58805d22f3b : SUCCESS in 5m 47s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/93da9bc5078d462fb76c472d898cc17b : SUCCESS in 48m 40s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/955e5e41120d460b8e7616dfa5e41ad5 : SUCCESS in 49m 38s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9b426b40faac4ed8b95eb794061cf49e : SUCCESS in 53m 01s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7aac458d01d245eb89f1b9af1cd21c56 : FAILURE in 39m 16s","accounts_in_message":[],"_revision_number":12},{"id":"8fa6c23aa22f5e1e21a51499a8f391ddebdc19b6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-25 16:29:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"6c95cbac699c0572133f17ceb5ceaa184c6457bd","revisions":{"1baee0722a6bfebed1a5e45a5016647537b97866":{"kind":"REWORK","_number":1,"created":"2020-09-09 10:00:33.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/14/750614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/1"}}},"commit":{"parents":[{"commit":"4a8a37346777650f6b6acd7d98f56b904548659e","subject":"Use gnocchi::db instead of database_connection in gnocchi class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4a8a37346777650f6b6acd7d98f56b904548659e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 10:00:14.000000000","tz":120},"subject":"DNM test with nfv sig repo added","message":"DNM test with nfv sig repo added\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1baee0722a6bfebed1a5e45a5016647537b97866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1baee0722a6bfebed1a5e45a5016647537b97866"}]},"branch":"refs/heads/master"},"50fb9c6f94ced5405aadfdfc9a79e9f73f243631":{"kind":"REWORK","_number":2,"created":"2020-09-16 08:01:04.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/14/750614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/2"}}},"commit":{"parents":[{"commit":"4a8a37346777650f6b6acd7d98f56b904548659e","subject":"Use gnocchi::db instead of database_connection in gnocchi class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4a8a37346777650f6b6acd7d98f56b904548659e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-16 08:01:05.000000000","tz":120},"subject":"DNM test with testing repo added","message":"DNM test with testing repo added\n\nload deps from https://trunk.rdoproject.org/centos8-master/delorean-deps-test.repo\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/50fb9c6f94ced5405aadfdfc9a79e9f73f243631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/50fb9c6f94ced5405aadfdfc9a79e9f73f243631"}]},"branch":"refs/heads/master"},"efc65420349980476fbf9b9d1987844bd3d3ea63":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-10-23 10:08:33.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/14/750614/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/3"}}},"commit":{"parents":[{"commit":"0555083a1f8ebed1ee47cc90bce1d48d4718f10a","subject":"Rename horizon_* ssl params to ssl_*","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0555083a1f8ebed1ee47cc90bce1d48d4718f10a"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-10-23 10:08:33.000000000","tz":0},"subject":"DNM test with testing repo added","message":"DNM test with testing repo added\n\nload deps from https://trunk.rdoproject.org/centos8-master/delorean-deps-test.repo\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/efc65420349980476fbf9b9d1987844bd3d3ea63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/efc65420349980476fbf9b9d1987844bd3d3ea63"}]},"branch":"refs/heads/master"},"d2fdd4c47a1ad22826f14cd8bbb00963baf360d6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-11-04 14:28:36.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/750614/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/4"}}},"commit":{"parents":[{"commit":"745214d0a7941038ff04775e63c339305464246b","subject":"Use magnum::keystone::keystone_auth for magnum config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/745214d0a7941038ff04775e63c339305464246b"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-11-04 14:28:36.000000000","tz":0},"subject":"DNM test with testing repo added","message":"DNM test with testing repo added\n\nload deps from https://trunk.rdoproject.org/centos8-master/delorean-deps-test.repo\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d2fdd4c47a1ad22826f14cd8bbb00963baf360d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d2fdd4c47a1ad22826f14cd8bbb00963baf360d6"}]},"branch":"refs/heads/master"},"6798b94a57dd2e685efafcf3a1b0b0970a74dbd0":{"kind":"REWORK","_number":5,"created":"2020-11-05 04:49:45.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/750614/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/5","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/5"}}},"commit":{"parents":[{"commit":"745214d0a7941038ff04775e63c339305464246b","subject":"Use magnum::keystone::keystone_auth for magnum config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/745214d0a7941038ff04775e63c339305464246b"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-11-05 04:49:40.000000000","tz":0},"subject":"DNM test with testing repo added","message":"DNM test with testing repo added\n\nload deps from https://trunk.rdoproject.org/centos8-master/delorean-deps-test.repo\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6798b94a57dd2e685efafcf3a1b0b0970a74dbd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6798b94a57dd2e685efafcf3a1b0b0970a74dbd0"}]},"branch":"refs/heads/master"},"9e93bfd380573d5f14dc21d3c8f1c14f4b3fd932":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-12-04 12:25:38.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/750614/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/6","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/6"}}},"commit":{"parents":[{"commit":"e3c80e1bc68b0429676c344232ca442a4ddbea48","subject":"Add newline to Nova migration secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e3c80e1bc68b0429676c344232ca442a4ddbea48"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-12-04 12:25:38.000000000","tz":0},"subject":"DNM test with testing repo added","message":"DNM test with testing repo added\n\nload deps from https://trunk.rdoproject.org/centos8-master/delorean-deps-test.repo\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9e93bfd380573d5f14dc21d3c8f1c14f4b3fd932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9e93bfd380573d5f14dc21d3c8f1c14f4b3fd932"}]},"branch":"refs/heads/master","description":"Rebase"},"5f2c8b4bef2ae359b86a33e5146861df0596d0b4":{"kind":"REWORK","_number":7,"created":"2020-12-11 11:56:13.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/750614/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/7","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/7"}}},"commit":{"parents":[{"commit":"e3c80e1bc68b0429676c344232ca442a4ddbea48","subject":"Add newline to Nova migration secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e3c80e1bc68b0429676c344232ca442a4ddbea48"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-11 11:55:49.000000000","tz":330},"subject":"DNM test with testing repo added","message":"DNM test with testing repo added\n\nload deps from https://trunk.rdoproject.org/centos8-master/delorean-deps-test.repo\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5f2c8b4bef2ae359b86a33e5146861df0596d0b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5f2c8b4bef2ae359b86a33e5146861df0596d0b4"}]},"branch":"refs/heads/master"},"8e8351a0d77fc780164a8d866bf3855a32608ea7":{"kind":"REWORK","_number":8,"created":"2020-12-15 06:31:42.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/750614/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/8","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/8"}}},"commit":{"parents":[{"commit":"5a6874c0ec103ee580c2f8bfb7b35c1f6fd7d99f","subject":"Move pci class invocation to before nova::compute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5a6874c0ec103ee580c2f8bfb7b35c1f6fd7d99f"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-15 06:31:25.000000000","tz":330},"subject":"DNM test with testing repo added","message":"DNM test with testing repo added\n\nload deps from https://trunk.rdoproject.org/centos8-master/delorean-deps-test.repo\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8e8351a0d77fc780164a8d866bf3855a32608ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8e8351a0d77fc780164a8d866bf3855a32608ea7"}]},"branch":"refs/heads/master"},"a04314475bee4e62acf77bb7de65fcc1d4917e1c":{"kind":"REWORK","_number":9,"created":"2021-02-05 07:11:10.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/750614/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/9","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/9"}}},"commit":{"parents":[{"commit":"70aaa2b6af9b9e4361c478b7a3d4c314060b67f2","subject":"Add CI jobs on CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/70aaa2b6af9b9e4361c478b7a3d4c314060b67f2"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-02-05 05:49:24.000000000","tz":330},"subject":"DNM test with testing repo added","message":"DNM test with testing repo added\n\nload deps from https://trunk.rdoproject.org/centos8-master/delorean-deps-test.repo\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a04314475bee4e62acf77bb7de65fcc1d4917e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a04314475bee4e62acf77bb7de65fcc1d4917e1c"}]},"branch":"refs/heads/master"},"c531a9a273ad5665dc3fda0aa19b4ce16cf563ef":{"kind":"REWORK","_number":10,"created":"2021-04-12 13:26:37.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/750614/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/10","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/10"}}},"commit":{"parents":[{"commit":"eabad465d8656c57a8af0f2612fc1f6cac618cfd","subject":"Drop CentOS8 based jobs for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eabad465d8656c57a8af0f2612fc1f6cac618cfd"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-04-12 13:25:55.000000000","tz":330},"subject":"DNM test with testing repo added","message":"DNM test with testing repo added\n\nload deps from https://trunk.rdoproject.org/centos8-master/delorean-deps-test.repo\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c531a9a273ad5665dc3fda0aa19b4ce16cf563ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c531a9a273ad5665dc3fda0aa19b4ce16cf563ef"}]},"branch":"refs/heads/master"},"affbb9cad8859b617bef8641bb62599f0babfc50":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-09-01 09:04:12.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/14/750614/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/11","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/11"}}},"commit":{"parents":[{"commit":"3bc4d7e722710d2f2aeac11560ea997dcf899169","subject":"Drop usage of cinder v2 params","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3bc4d7e722710d2f2aeac11560ea997dcf899169"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-09-01 09:04:12.000000000","tz":0},"subject":"DNM test with testing repo added","message":"DNM test with testing repo added\n\nload deps from https://trunk.rdoproject.org/centos8-master/delorean-deps-test.repo\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/affbb9cad8859b617bef8641bb62599f0babfc50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/affbb9cad8859b617bef8641bb62599f0babfc50"}]},"branch":"refs/heads/master","description":"Rebase"},"6c95cbac699c0572133f17ceb5ceaa184c6457bd":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-09-01 12:03:06.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/14/750614/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/750614/12","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/750614/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/750614/12"}}},"commit":{"parents":[{"commit":"091b810f1ad9de6e1fdf584fdea84e16d9ad858b","subject":"Move puppet-rsync to main branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/091b810f1ad9de6e1fdf584fdea84e16d9ad858b"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-09 09:59:42.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-09-01 12:03:06.000000000","tz":0},"subject":"DNM test with testing repo added","message":"DNM test with testing repo added\n\nload deps from https://trunk.rdoproject.org/centos8-master/delorean-deps-test.repo\n\nChange-Id: I9c29d0495bc4665595ebfa728b227cc654ada6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6c95cbac699c0572133f17ceb5ceaa184c6457bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6c95cbac699c0572133f17ceb5ceaa184c6457bd"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
