)]}'
{"id":"x%2Fpackstack~766691","triplet_id":"x%2Fpackstack~master~Ic10b461dde012e01de87eaf03745f17fbaf704e8","project":"x/packstack","branch":"master","topic":"distro_detect","hashtags":[],"change_id":"Ic10b461dde012e01de87eaf03745f17fbaf704e8","subject":"Rely on ID/VERSION_ID to detect distro","status":"MERGED","created":"2020-12-11 10:19:43.000000000","updated":"2020-12-15 13:39:02.000000000","submitted":"2020-12-15 13:39:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"766691-distro_detect","meta_rev_id":"aea8cc15d17759360b4e16ba51b996e1d280c05b","_number":766691,"virtual_id_number":766691,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-15 13:39:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2020-12-15 11:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-14 15:14:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2020-12-15 11:53:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-11 10:43:28.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-12-11 10:46:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-14 15:14:42.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-12-15 11:29:04.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REMOVED"},{"updated":"2020-12-15 11:29:18.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"957dc1859e690fae632b9f8d588fa3118727da12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-11 10:19:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52c59240efa9cfa34611f017db15dd694b1de270","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-12-11 10:43:28.000000000","message":"Patch Set 1: Code-Review+2\n\nI guess we will need to backport this to all stable branches, right?","accounts_in_message":[],"_revision_number":1},{"id":"ad5632ffce49225d73c3284b18810b10c2612397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 10:46:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f5ce7c19ced465e97fc0dc50d23e624 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f309b16d6fdd482eabc9bdd213d43f2f : SUCCESS in 4m 05s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/44f2d86ec5ee43d3ad4cc138043b863a : POST_FAILURE in 15m 17s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/763258cce4814075918e02d326345e97 : POST_FAILURE in 14m 37s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2c8b328f167e4f14ae46085d038748bb : POST_FAILURE in 13m 59s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/98799a2d5fb64466a5526cd4d458993d : POST_FAILURE in 20m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f7203df743544293cf9e7d8b5a8603d0ed639a8a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-11 10:53:43.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+2\n\u003e \n\u003e I guess we will need to backport this to all stable branches, right?\n\nyeap till train.","accounts_in_message":[],"_revision_number":1},{"id":"cb792e6ac9f3f45839f50d2f4a13fca18369f309","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-12-11 10:57:10.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2130bcbabfa8e9a65473a190eb99c818cae76386","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-11 11:02:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"528e1683ca8620cd13cac1339334eea6877cc762","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-12-11 11:05:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25db21279243ae67196df98b3ddbad5f6de4da9e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-11 11:38:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e19c84917b50146cc6dfd6f4fd7eccb0ee8829b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-14 12:51:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7a1b2d10fcded1498bb171fd9efdbaa7a2cce5cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 14:13:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4e93d777cd4a8f8f426c8e85a4b21e : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d73f14c563514e99b87a90bd8e2bf5b1 : SUCCESS in 3m 25s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a429a080ab8642b1aaf60ee7cacd35a0 : SUCCESS in 55m 42s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1a10aa5d01c64775b08c1c76128fe607 : SUCCESS in 42m 45s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e4c0f550592c46ae93a391013500059c : SUCCESS in 50m 50s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/788ca214a54743829bc6b960706454ce : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":2},{"id":"3d3ca62d963b413b8d261c2554bb1d719a6d3a39","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-12-14 15:14:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c5ceb23c9f851376317119c5c58ad8d7cf93149b","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-15 11:29:04.000000000","message":"Removed reviewer Alfredo Moralejo.","accounts_in_message":[],"_revision_number":2},{"id":"a037b977d7ef849457db6fc6be969e1c86d7a8a4","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-12-15 11:53:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f91678b79c8754aec8d736ab708bde82dc0bf456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 11:53:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e8104f52f25490307bf9b8c3cd4191c0894d01ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 13:39:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b70f4b684d64582a4f40c9d65c06b15 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36f1796261284ce7a5e197b9ad63ae01 : SUCCESS in 3m 38s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/242fae6488ae40bba92b7086b1bd54c5 : SUCCESS in 53m 23s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/92ad95e4db5242e0a3e750e7488420a4 : SUCCESS in 55m 12s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/21db93909bcb44e8a10867547c737d79 : SUCCESS in 57m 06s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9fe9656dbc9242cf9688c36762e8d60b : SUCCESS in 1h 34m 47s","accounts_in_message":[],"_revision_number":2},{"id":"aea8cc15d17759360b4e16ba51b996e1d280c05b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 13:39:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"026d2feefd367aa9709b56e3f24c50c8420dc4de","revisions":{"045a7231bcbfe89378b470ca44403653efdd73fd":{"kind":"REWORK","_number":1,"created":"2020-12-11 10:19:43.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/91/766691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/91/766691/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/91/766691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/91/766691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/91/766691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/91/766691/1"}}},"commit":{"parents":[{"commit":"e3c80e1bc68b0429676c344232ca442a4ddbea48","subject":"Add newline to Nova migration secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e3c80e1bc68b0429676c344232ca442a4ddbea48"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-11 08:22:06.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-11 08:22:13.000000000","tz":330},"subject":"Rely on ID/VERSION_ID to detect distro","message":"Rely on ID/VERSION_ID to detect distro\n\nREDHAT_SUPPORT_PRODUCT* are removed from recent\nCentOS8.3 release[1], Let\u0027s use ID/VERSION_ID instead\nwhich are available[2] and would serve our purpose.\n\n[1] https://git.centos.org/rpms/centos-linux-release/c/2e3df0\n[2] https://www.freedesktop.org/software/systemd/man/os-release.html\n\nChange-Id: Ic10b461dde012e01de87eaf03745f17fbaf704e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/045a7231bcbfe89378b470ca44403653efdd73fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/045a7231bcbfe89378b470ca44403653efdd73fd"}]},"branch":"refs/heads/master"},"026d2feefd367aa9709b56e3f24c50c8420dc4de":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-14 12:51:27.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/91/766691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/91/766691/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/91/766691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/91/766691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/91/766691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/91/766691/2"}}},"commit":{"parents":[{"commit":"5a6874c0ec103ee580c2f8bfb7b35c1f6fd7d99f","subject":"Move pci class invocation to before nova::compute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5a6874c0ec103ee580c2f8bfb7b35c1f6fd7d99f"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-11 08:22:06.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-12-14 12:51:27.000000000","tz":0},"subject":"Rely on ID/VERSION_ID to detect distro","message":"Rely on ID/VERSION_ID to detect distro\n\nREDHAT_SUPPORT_PRODUCT* are removed from recent\nCentOS8.3 release[1], Let\u0027s use ID/VERSION_ID instead\nwhich are available[2] and would serve our purpose.\n\n[1] https://git.centos.org/rpms/centos-linux-release/c/2e3df0\n[2] https://www.freedesktop.org/software/systemd/man/os-release.html\n\nChange-Id: Ic10b461dde012e01de87eaf03745f17fbaf704e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/026d2feefd367aa9709b56e3f24c50c8420dc4de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/026d2feefd367aa9709b56e3f24c50c8420dc4de"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
