)]}'
{"id":"x%2Fpackstack~767055","triplet_id":"x%2Fpackstack~stable%2Fvictoria~Ic10b461dde012e01de87eaf03745f17fbaf704e8","project":"x/packstack","branch":"stable/victoria","topic":"distro_detect-stable/victoria","hashtags":[],"change_id":"Ic10b461dde012e01de87eaf03745f17fbaf704e8","subject":"Rely on ID/VERSION_ID to detect distro","status":"MERGED","created":"2020-12-15 13:40:36.000000000","updated":"2020-12-16 11:40:20.000000000","submitted":"2020-12-16 11:40:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767055-distro_detect-stable/victoria","cherry_pick_of_change":766691,"cherry_pick_of_patch_set":2,"meta_rev_id":"eef45fb589fab1c4d53afd789518a53e93745f7a","_number":767055,"virtual_id_number":767055,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-16 11:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2020-12-16 09:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-16 11:40:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2020-12-16 09:22:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 13:43:03.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-12-15 14:11:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-16 08:58:58.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"8a8a64301927c9069f045ed5e89510e9f24977c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-15 13:40:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ec117020d00b4f9e243afef4509905fce03d1747","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-12-15 13:43:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ad7da762555c4ec62e1bb191dd12b506345b181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 14:11:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/714fe02217c240d0a0eac6ec6256cdf2 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c15dbd8cce3414fa1e7e31bcb8155fb : SUCCESS in 5m 09s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/63c0ca522aa54ff183ba300005439767 : POST_FAILURE in 17m 58s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7f0bc130e0ea4ca99187620dd0b097d0 : POST_FAILURE in 19m 49s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b939c6335c6b448aa02be9f921c31446 : POST_FAILURE in 11m 25s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/53b5b1c50e7b492f850079d0bc8cf5b9 : POST_FAILURE in 23m 36s","accounts_in_message":[],"_revision_number":1},{"id":"42c314564822a0f35f50653783dd0c8fd5effce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-16 06:55:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"123c0bd4a2a027b9fa879c4d4db78729a513c74d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 08:44:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca5f1b9963f447fa929e5dc769375ed : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1961beaa830145fdaa25bb0f1b0530b0 : SUCCESS in 4m 08s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5e0ef6612d534e50bc43266b359609df : SUCCESS in 46m 08s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/60bb9b00df154ddf922cb57a830bde95 : SUCCESS in 44m 26s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9f432eded9d341499d85bb3c98137faf : SUCCESS in 45m 35s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c52c623284ee463c86bf330eccddf947 : SUCCESS in 51m 07s","accounts_in_message":[],"_revision_number":2},{"id":"2b7fc993f7a2dbcf3864ad0cdc0b229425823bc9","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-12-16 09:22:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f5868a66c63f7c46ecc528641fa9623f1b6384dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 10:05:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"855ed38f168d92fa8d2e632a629b334a8a0c5bb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 11:40:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5fda0e2907e48a997d8743e82f1f670 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ae90f3a5f344cbdb3bac6fbf854b889 : SUCCESS in 6m 35s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d23fb5407b864e5db3aa75f70e6b7dbb : SUCCESS in 49m 22s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/115dd43364f649b787351ca802145708 : SUCCESS in 46m 52s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c28239015fa249b893088904040bf521 : SUCCESS in 56m 11s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6268525c83b247c694717980a912f23e : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":2},{"id":"5831961c30a7d53f0314692258a8a884c0b455ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 11:40:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7d1a8f3c41cc3898786ef5681a552a3a5adedfde","revisions":{"97c9d374f61cdfef48c0b373483f2788cfcf8e0b":{"kind":"REWORK","_number":1,"created":"2020-12-15 13:40:36.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/55/767055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/55/767055/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/55/767055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/55/767055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/55/767055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/55/767055/1"}}},"commit":{"parents":[{"commit":"7ef539bc37a02c80c6ad47a1b875619b45bac3d5","subject":"Use netifaces instead of facter to find ip address","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7ef539bc37a02c80c6ad47a1b875619b45bac3d5"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-11 08:22:06.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-12-15 13:40:36.000000000","tz":0},"subject":"Rely on ID/VERSION_ID to detect distro","message":"Rely on ID/VERSION_ID to detect distro\n\nREDHAT_SUPPORT_PRODUCT* are removed from recent\nCentOS8.3 release[1], Let\u0027s use ID/VERSION_ID instead\nwhich are available[2] and would serve our purpose.\n\n[1] https://git.centos.org/rpms/centos-linux-release/c/2e3df0\n[2] https://www.freedesktop.org/software/systemd/man/os-release.html\n\nChange-Id: Ic10b461dde012e01de87eaf03745f17fbaf704e8\n(cherry picked from commit 026d2feefd367aa9709b56e3f24c50c8420dc4de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/97c9d374f61cdfef48c0b373483f2788cfcf8e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/97c9d374f61cdfef48c0b373483f2788cfcf8e0b"}]},"branch":"refs/heads/stable/victoria"},"7d1a8f3c41cc3898786ef5681a552a3a5adedfde":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-16 06:55:16.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/55/767055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/55/767055/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/55/767055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/55/767055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/55/767055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/55/767055/2"}}},"commit":{"parents":[{"commit":"20d8eb9d439f0ebd59b3797514363f047df147fc","subject":"Install python3-wheel to unblock CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/20d8eb9d439f0ebd59b3797514363f047df147fc"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-11 08:22:06.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-12-16 06:55:16.000000000","tz":0},"subject":"Rely on ID/VERSION_ID to detect distro","message":"Rely on ID/VERSION_ID to detect distro\n\nREDHAT_SUPPORT_PRODUCT* are removed from recent\nCentOS8.3 release[1], Let\u0027s use ID/VERSION_ID instead\nwhich are available[2] and would serve our purpose.\n\n[1] https://git.centos.org/rpms/centos-linux-release/c/2e3df0\n[2] https://www.freedesktop.org/software/systemd/man/os-release.html\n\nChange-Id: Ic10b461dde012e01de87eaf03745f17fbaf704e8\n(cherry picked from commit 026d2feefd367aa9709b56e3f24c50c8420dc4de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7d1a8f3c41cc3898786ef5681a552a3a5adedfde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7d1a8f3c41cc3898786ef5681a552a3a5adedfde"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
