)]}'
{"id":"x%2Fpackstack~767137","triplet_id":"x%2Fpackstack~stable%2Ftrain~Ic10b461dde012e01de87eaf03745f17fbaf704e8","project":"x/packstack","branch":"stable/train","topic":"distro_detect-stable/victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"Ic10b461dde012e01de87eaf03745f17fbaf704e8","subject":"Rely on ID/VERSION_ID to detect distro","status":"MERGED","created":"2020-12-15 13:41:21.000000000","updated":"2020-12-17 11:41:00.000000000","submitted":"2020-12-17 11:41:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767137-distro_detect-stable/victoria-stable/ussuri-stable/train","cherry_pick_of_change":767136,"cherry_pick_of_patch_set":1,"meta_rev_id":"6ee8953df44309b4ffc5d5bbd18ea7eaba032b3f","_number":767137,"virtual_id_number":767137,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-17 11:40:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2020-12-17 08:35:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-16 15:53:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2020-12-17 08:35:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 13:42:55.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-12-15 14:50:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-16 15:53:49.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"bf7fbccd9826bb92568c067c79cd0b91cc4735ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-15 13:41:21.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"9bef88dd60567d57d4de51d23c6474bdc19bb492","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-12-15 13:42:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"750fc5ce9670efe1054cce5af7ffdcc752339c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 14:50:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe1f7a69229643408bf4d8b6770f68dc : FAILURE in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f853c6747a324a58bc041d3db6d5e281 : SUCCESS in 4m 50s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/309737fbd04748a2a31d5a6ef8cf14cb : SUCCESS in 50m 30s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4307b8424fbf4cd5a07e38ac4a3ce9ac : SUCCESS in 45m 12s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/1d113aa094044bf7bee8f6c1a2788dea : SUCCESS in 51m 19s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/773c13efe8374b3688d9bde3a4d53f0b : SUCCESS in 1h 00m 58s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/317cdfbc2c1f4a2b9fd9ed07f586f191 : SUCCESS in 51m 29s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1f84ac134163413abb1ece1a36049617 : SUCCESS in 39m 38s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ace4991913f74071b88eb1968286e66f : SUCCESS in 45m 53s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1377c601fa244b7db3bdad39bd7e7207 : SUCCESS in 56m 13s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a57589a8bbb4179826d4c27950555dc : FAILURE in 3m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"888367fdbdc41a122d764043590cedcdfd09fa96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-16 07:03:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"afc4e0ca086139f4c3fcc450050688f54f8da927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 10:06:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24daa133407440369293b934c554f8ee : FAILURE in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/361d2ae7cf59490a8b2623fc763cabd7 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2fc2639e39d4abe8cf625ffe49131b0 : SUCCESS in 4m 00s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/6bca8503f932490d9aa9125ad1ff3faa : SUCCESS in 49m 12s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a685e3198a2d4b2196ea247c57a74f5b : SUCCESS in 39m 29s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/7b53183baefc4f5aa0b026b131e5d551 : SUCCESS in 40m 45s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e815eef7869943ed9936c7d806e61c8d : SUCCESS in 40m 18s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4f01b61b609344f5807a539beaf70e20 : SUCCESS in 44m 48s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/186ff2df6558457ca2e3edb7405fd645 : SUCCESS in 44m 54s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ae9e470143a34c998ca6db8a7fe91eb7 : SUCCESS in 47m 06s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9776e1b9b21747a1bf5e883aaf793b46 : SUCCESS in 49m 53s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba58b36bc63d435599d2b7f408b41d8d : FAILURE in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9282cbdd3df4efd44151c2cd5b0db2304a138b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-16 14:28:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1af879150d4983921d8fa207f77a67f708c482e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 15:40:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83b5a0b5b9524b9f92e7c01b533ebbd5 : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/692ded9a9d9345daa45f13ebe62d8e89 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e43337615bf3425f8cb41d5ff04790f3 : SUCCESS in 4m 27s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/32b0bbbddb844712b81d3f787e832209 : SUCCESS in 40m 42s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/69d3aa3c2cf24ce890bf1dfc9b059e01 : SUCCESS in 40m 43s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/e06358c12e42483eacee4003bf39f32d : SUCCESS in 56m 37s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/656cf39ad3824a78a7ea4effd902dd86 : SUCCESS in 43m 01s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/62e451eafcd24de8ab5d76afff768666 : SUCCESS in 43m 09s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/281fcb709a93443898c94719a21f6831 : SUCCESS in 48m 40s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9c431916b72c42e0b3ccb466ccc8da9b : SUCCESS in 43m 09s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f8e9f09b77434de4adf64b727fa07e92 : SUCCESS in 42m 19s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1c760038fdd4f15afded82c0c796374 : SUCCESS in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5e82901af9ad2eb3bb0f349106c498d9eaf9a26c","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-12-16 15:53:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"93b54c7a3d30fe88d7b82ef05e8c987fbe5df670","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-12-17 08:35:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"505043ba06f569e9edf6cff802a9f49bda31832d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 08:36:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a4f8c378c89a97b344afd49ef180aaf6f6f924e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 09:36:56.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a95de781d835438993930c6a48547836 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/caffbc90a47549f2bea958e29ad452ac : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36f69389d4a045eea8c666fa286fbf7e : SUCCESS in 3m 18s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/0204aeb8db30418ab4e12acf76a42130 : SUCCESS in 38m 33s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/26604bef3ff948eaace4590b2fa37813 : SUCCESS in 48m 07s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/a6fd0b82566a4b38b381192dc45f46c4 : SUCCESS in 51m 13s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/19ecb10b0c8c4b5ca2fc233548c26f1a : FAILURE in 13m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9660b567781b969cc1de99fc0b9d30a90c8bfeca","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-17 09:45:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"96b31f06a2c6cddc818897a79d276525451f8bb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 10:53:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4577f5fc7b244673a78e0a8962eb388e : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4edde52e9c72425586b50390cebb5d1c : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65d0e289e96742c080ef86982046534c : SUCCESS in 3m 29s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/689d6f890f2a47f2babc4fa0fd888580 : SUCCESS in 42m 43s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/64c8643c109447aba24ce7ff6d3197d3 : SUCCESS in 42m 29s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/66fa82f76a2b414b9eabd57dadf2a7a0 : SUCCESS in 53m 37s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f502321989b740429cd653d3f1441915 : SUCCESS in 1h 01m 49s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/3aa1d67da9104c11a18bb7a7be14e7fc : SUCCESS in 41m 48s (non-voting)\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/aa68323555334173b36f9975403b1bfd : FAILURE in 1h 02m 15s (non-voting)\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b178fb8474ad4b2099471ea387cc9b67 : FAILURE in 41m 05s (non-voting)\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ff7f96f8f6c64848872139ba01328a03 : SUCCESS in 41m 09s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19b01696642b4efc8d40c6018e01ddfe : SUCCESS in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6c38d88f2f7777a0ca9c6fc2601dd3c927175fdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 10:53:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"78b1705d89117c01ee193e7b30d8dea93bba6dfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 11:40:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31c0727bb7a24a8bb593a4cbfdbc23fd : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a036fb6daaa4a2194ed928897e4963a : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d73c813bd1af4912a3d1467f88b63fe8 : SUCCESS in 3m 42s\n- packstack-integration-scenario001-tempest https://zuul.opendev.org/t/openstack/build/dffee886216642a6bd7168d13191ac99 : SUCCESS in 40m 31s\n- packstack-integration-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7fdd78dc734f4eb2b104399824d69160 : SUCCESS in 40m 16s\n- packstack-integration-scenario003-tempest https://zuul.opendev.org/t/openstack/build/7e6560e3718b472cb24df32830534f0f : SUCCESS in 38m 54s\n- packstack-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9e8e14d70f3946989d3d7c22783ac73b : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":3},{"id":"6ee8953df44309b4ffc5d5bbd18ea7eaba032b3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 11:41:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ceead5c48e339a9b4642d6451f0f1b2f95b624c6","revisions":{"dc71a35eb6a191f7ea7a70d0c856c38fa663920f":{"kind":"REWORK","_number":1,"created":"2020-12-15 13:41:21.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/37/767137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/37/767137/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/37/767137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/37/767137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/37/767137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/37/767137/1"}}},"commit":{"parents":[{"commit":"12c99860b345cc05610eb7ac19211e89429ad345","subject":"Merge \"Add jobs for CentOS 8\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/12c99860b345cc05610eb7ac19211e89429ad345"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-11 08:22:06.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-12-15 13:41:21.000000000","tz":0},"subject":"Rely on ID/VERSION_ID to detect distro","message":"Rely on ID/VERSION_ID to detect distro\n\nREDHAT_SUPPORT_PRODUCT* are removed from recent\nCentOS8.3 release[1], Let\u0027s use ID/VERSION_ID instead\nwhich are available[2] and would serve our purpose.\n\n[1] https://git.centos.org/rpms/centos-linux-release/c/2e3df0\n[2] https://www.freedesktop.org/software/systemd/man/os-release.html\n\nChange-Id: Ic10b461dde012e01de87eaf03745f17fbaf704e8\n(cherry picked from commit 026d2feefd367aa9709b56e3f24c50c8420dc4de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/dc71a35eb6a191f7ea7a70d0c856c38fa663920f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/dc71a35eb6a191f7ea7a70d0c856c38fa663920f"}]},"branch":"refs/heads/stable/train"},"d7d7336e7b7dcf4612b8db3f56cd3e01b4ba5681":{"kind":"REWORK","_number":2,"created":"2020-12-16 07:03:24.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/37/767137/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/37/767137/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/37/767137/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/37/767137/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/37/767137/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/37/767137/2"}}},"commit":{"parents":[{"commit":"12c99860b345cc05610eb7ac19211e89429ad345","subject":"Merge \"Add jobs for CentOS 8\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/12c99860b345cc05610eb7ac19211e89429ad345"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-11 08:22:06.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-16 07:02:57.000000000","tz":330},"subject":"Rely on ID/VERSION_ID to detect distro","message":"Rely on ID/VERSION_ID to detect distro\n\nREDHAT_SUPPORT_PRODUCT* are removed from recent\nCentOS8.3 release[1], Let\u0027s use ID/VERSION_ID instead\nwhich are available[2] and would serve our purpose.\n\n[1] https://git.centos.org/rpms/centos-linux-release/c/2e3df0\n[2] https://www.freedesktop.org/software/systemd/man/os-release.html\n\nAlso switch to train constraints.\n\nChange-Id: Ic10b461dde012e01de87eaf03745f17fbaf704e8\n(cherry picked from commit 026d2feefd367aa9709b56e3f24c50c8420dc4de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d7d7336e7b7dcf4612b8db3f56cd3e01b4ba5681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d7d7336e7b7dcf4612b8db3f56cd3e01b4ba5681"}]},"branch":"refs/heads/stable/train"},"ceead5c48e339a9b4642d6451f0f1b2f95b624c6":{"kind":"REWORK","_number":3,"created":"2020-12-16 14:28:30.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/37/767137/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/37/767137/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/37/767137/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/37/767137/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/37/767137/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/37/767137/3"}}},"commit":{"parents":[{"commit":"12c99860b345cc05610eb7ac19211e89429ad345","subject":"Merge \"Add jobs for CentOS 8\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/12c99860b345cc05610eb7ac19211e89429ad345"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-11 08:22:06.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-16 14:27:27.000000000","tz":330},"subject":"Rely on ID/VERSION_ID to detect distro","message":"Rely on ID/VERSION_ID to detect distro\n\nREDHAT_SUPPORT_PRODUCT* are removed from recent\nCentOS8.3 release[1], Let\u0027s use ID/VERSION_ID instead\nwhich are available[2] and would serve our purpose.\n\n[1] https://git.centos.org/rpms/centos-linux-release/c/2e3df0\n[2] https://www.freedesktop.org/software/systemd/man/os-release.html\n\nAlso switch to train constraints and remove cap for Sphinx\nand let it use from constraints which is 3.0.1 currently.\n\nChange-Id: Ic10b461dde012e01de87eaf03745f17fbaf704e8\n(cherry picked from commit 026d2feefd367aa9709b56e3f24c50c8420dc4de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ceead5c48e339a9b4642d6451f0f1b2f95b624c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ceead5c48e339a9b4642d6451f0f1b2f95b624c6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
