)]}'
{"id":"x%2Fpackstack~767167","triplet_id":"x%2Fpackstack~stable%2Fvictoria~I264a0104aaf9ad8b927126cb523a4e8df9c5e6b7","project":"x/packstack","branch":"stable/victoria","hashtags":[],"change_id":"I264a0104aaf9ad8b927126cb523a4e8df9c5e6b7","subject":"Install python3-wheel to unblock CI jobs","status":"MERGED","created":"2020-12-15 14:47:12.000000000","updated":"2020-12-16 11:40:17.000000000","submitted":"2020-12-16 11:40:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767167","meta_rev_id":"957949987700b42576d07a56170e0f8c9f663240","_number":767167,"virtual_id_number":767167,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-16 11:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2020-12-16 10:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-16 08:58:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2020-12-16 10:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 14:47:12.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-12-15 15:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-16 08:47:08.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2020-12-16 08:47:08.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"}],"messages":[{"id":"8d0875e8b2afce401e00fad2cee0475cbcbe2cb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-15 14:47:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e12148673ca09275d469162a4968601ef920c531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 15:23:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/124b8968ee1844e5a223a967ccdcbb25 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9111e3e92ac0462eaaf7fe1cb181cb3a : SUCCESS in 3m 36s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/305bc326389a413794ef52fc18ac1649 : FAILURE in 20m 27s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b5d0e3fd745e42329867fad5f2b6dde9 : FAILURE in 12m 42s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/06badc6df2ff4f6e9476fce87f39fbd3 : FAILURE in 23m 40s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/67a2d9b6149f47378e2ccfca02c58e3e : FAILURE in 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7ab8702c8ebbf329e513891864f7f5a61017e67f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-16 05:46:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a7037fb272b265627a37e4e900724c68310abb26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 07:48:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48450d6aab874e32b357ded50425730c : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa1c96abf3ed4765bbc0cdf2e0f17aa2 : SUCCESS in 3m 19s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1b9d8986e0c3426697d8cfe25b24c70e : SUCCESS in 53m 30s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/07f57b35e14842c09f537fb0c81e8aa4 : SUCCESS in 49m 34s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6f61912cd1b14360ac0febe8e39e71b9 : SUCCESS in 53m 37s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/99288b5817a44c6cb7a9c4a0f49b7dc1 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":2},{"id":"4b8fa72cbe6e48ecb1795bf69b9fd27a89246181","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-12-16 08:58:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c6b55d4e62bf7dca11727ffb7285b64fd15a4b6","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-12-16 10:04:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"78a21f05b09f69c4285665def3b9076838032417","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 10:05:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"622ad47f0649950b65d1a91d0718c61ecf7391ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 11:40:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66c6e257b21140f080ee4b54158800d3 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bd85bcf145a41a58bbfd96a6025ce04 : SUCCESS in 3m 21s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b7598362854740a6a751fe273af5b701 : SUCCESS in 41m 18s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/36105735e96c4055aea0518010ed68cc : SUCCESS in 44m 17s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b52d0952402d46e8aa80e7053806762e : SUCCESS in 47m 00s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5987cf3e61ce4bc39aa0b549fceba917 : SUCCESS in 1h 33m 00s","accounts_in_message":[],"_revision_number":2},{"id":"957949987700b42576d07a56170e0f8c9f663240","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 11:40:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"20d8eb9d439f0ebd59b3797514363f047df147fc","revisions":{"0ee20b39ac93392389eb2abdf0e269de45554a2b":{"kind":"REWORK","_number":1,"created":"2020-12-15 14:47:12.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/67/767167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/67/767167/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/67/767167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/67/767167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/67/767167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/67/767167/1"}}},"commit":{"parents":[{"commit":"7ef539bc37a02c80c6ad47a1b875619b45bac3d5","subject":"Use netifaces instead of facter to find ip address","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7ef539bc37a02c80c6ad47a1b875619b45bac3d5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-11-10 09:23:05.000000000","tz":60},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-15 14:42:43.000000000","tz":330},"subject":"Install python3-wheel to unblock CI jobs","message":"Install python3-wheel to unblock CI jobs\n\nThis is partial cherry-pick from master commit to install\npython3-wheel to unblock CI jobs which are failing with error:\n\n\"ERROR: \u0027pip wheel\u0027 requires the \u0027wheel\u0027 package. To fix this, run: pip install wheel\"\n\nChange-Id: I264a0104aaf9ad8b927126cb523a4e8df9c5e6b7\n(cherry picked from commit 19c71e92d1455a5d981db178ff6de8e68b2ed7db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0ee20b39ac93392389eb2abdf0e269de45554a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0ee20b39ac93392389eb2abdf0e269de45554a2b"}]},"branch":"refs/heads/stable/victoria"},"20d8eb9d439f0ebd59b3797514363f047df147fc":{"kind":"REWORK","_number":2,"created":"2020-12-16 05:46:11.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/67/767167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/67/767167/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/67/767167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/67/767167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/67/767167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/67/767167/2"}}},"commit":{"parents":[{"commit":"7ef539bc37a02c80c6ad47a1b875619b45bac3d5","subject":"Use netifaces instead of facter to find ip address","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7ef539bc37a02c80c6ad47a1b875619b45bac3d5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-11-10 09:23:05.000000000","tz":60},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-12-16 05:45:05.000000000","tz":330},"subject":"Install python3-wheel to unblock CI jobs","message":"Install python3-wheel to unblock CI jobs\n\nThis is partial cherry-pick from master commit to install\npython3-wheel to unblock CI jobs which are failing with error:\n\n\"ERROR: \u0027pip wheel\u0027 requires the \u0027wheel\u0027 package. To fix this, run: pip install wheel\"\n\nAlso switch to victoria and victoria current-passed-ci promotion\nrepos.\n\nChange-Id: I264a0104aaf9ad8b927126cb523a4e8df9c5e6b7\n(cherry picked from commit 19c71e92d1455a5d981db178ff6de8e68b2ed7db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/20d8eb9d439f0ebd59b3797514363f047df147fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/20d8eb9d439f0ebd59b3797514363f047df147fc"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
