)]}'
{"id":"x%2Fpackstack~780546","triplet_id":"x%2Fpackstack~master~Ieda2cd0e7d3f25b99c28d5b338ea45f84b8a8a38","project":"x/packstack","branch":"master","topic":"c8s","hashtags":[],"change_id":"Ieda2cd0e7d3f25b99c28d5b338ea45f84b8a8a38","subject":"Drop CentOS8 based jobs for Wallaby","status":"MERGED","created":"2021-03-15 11:06:42.000000000","updated":"2021-04-06 14:20:04.000000000","submitted":"2021-04-06 14:20:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"780546-c8s","meta_rev_id":"af1d82aa4ae1d2ec40635273ef4c3c5d2ef0ca3f","_number":780546,"virtual_id_number":780546,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 14:20:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2021-04-06 13:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-06 07:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2021-04-06 13:07:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 12:19:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-16 09:15:09.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2021-03-16 09:15:21.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2021-03-16 11:10:25.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"ed3afafd8cf2d1738f08bac216ecc0ed4109012f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-03-15 11:06:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f370507a1be3eceffe8255149c5b21f681644ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 12:19:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4311920127142179ddd0e03c52477ab : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25ad7c6b1b4444bea8f9b3866cf25b91 : SUCCESS in 4m 18s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/feb736e8492847c68295d6a59ac01dab : SUCCESS in 51m 10s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/03f1aa48b2a140cd8c90c151a0cc767e : SUCCESS in 39m 08s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7e90cbfa3a0e490e84d7fac788c00355 : SUCCESS in 55m 33s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c4b550771d6a41238902d868bbdf5ad0 : SUCCESS in 1h 06m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b36e7908ea7ed3d76df99796f1d2684ee73aae88","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-03-16 09:15:09.000000000","message":"Patch Set 1:\n\nMay we keep them as non-voting while c8 is supported?, it\u0027s hard to let an old friend go... 😊","accounts_in_message":[],"_revision_number":1},{"id":"5a2929323bba46a9dc0370e4b0490db28fdb6150","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-03-16 09:18:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"48dad2cd156dd8f844e2bdadfe1c7d4a17316692","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-03-16 10:11:12.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)\n\n\u003e Patch Set 1:\n\u003e \n\u003e May we keep them as non-voting while c8 is supported?, it\u0027s hard to let an old friend go... 😊\n\nNot sure if that helps to keep it as non-voting for wallaby+, what i thought was to drop centos8 jobs from both packstack and poi master and then drop from puppet-promotion pipeline keeping only centos8-stream jobs for wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"150ce89137837a9b76661f2fa3cbcb9a211fd86c","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-03-16 11:10:25.000000000","message":"Patch Set 1: Code-Review+2\n\n\u003e Patch Set 1: Workflow-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e May we keep them as non-voting while c8 is supported?, it\u0027s hard to let an old friend go... 😊\n\u003e \n\u003e Not sure if that helps to keep it as non-voting for wallaby+, what i thought was to drop centos8 jobs from both packstack and poi master and then drop from puppet-promotion pipeline keeping only centos8-stream jobs for wallaby.\n\nok, let\u0027s do it.","accounts_in_message":[],"_revision_number":1},{"id":"c71160d0bb9404a32c12b9ffbc85cceeb95e2fd5","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-03-16 11:10:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0abd4ded0483daf5e3643082c60344574170c046","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-03-31 04:45:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dca41c7c645a500c9d9526a768db8b18291e4f19","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-04-06 06:46:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"228d7b2fc0142c2b78cf7a9144db948e4a6850a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-04-06 06:46:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"afc08d4262a8a1f30f2e0c18c49285af43454089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 07:51:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0d654cfd5e24e1d8eaf07d6df3c2879 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e93fff706334e449105f63d349b25ca : SUCCESS in 4m 26s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c427c39b2df842f9875406deb9b0f29f : SUCCESS in 47m 45s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6c2d312b6a094c4dac25e3885f357914 : SUCCESS in 35m 06s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/bdf1181348574a009deb03a9627ea5dd : SUCCESS in 54m 53s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/314ea8d2a2d14d80874494894c999f40 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":2},{"id":"4d956762085b269c0285e336d047b634ceeb0657","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-04-06 07:53:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"89da787ba549f74d1d78421d0ccc548482b431e0","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-04-06 13:07:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c75a4a88d19a1c917ffb865227aa721515ba9bf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:10:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"909ed4bb1e6dc23059e8a997ea940bab55ccbe55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 14:20:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d1150f8f0ac4281a2711869186cc66c : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c802f7da1d294f18a6ac9705539d26f5 : SUCCESS in 6m 45s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8438df05cf494eaa9a8a3a365c45ee16 : SUCCESS in 51m 46s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/34dc40e695d54286a23595c1d6de3e35 : SUCCESS in 47m 15s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/4faf4d29488b49db8b7a020636bf36af : SUCCESS in 1h 03m 05s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/18c44e6b5e5743ae8f0158c18839d5ad : SUCCESS in 50m 38s","accounts_in_message":[],"_revision_number":2},{"id":"af1d82aa4ae1d2ec40635273ef4c3c5d2ef0ca3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 14:20:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eabad465d8656c57a8af0f2612fc1f6cac618cfd","revisions":{"2b19cc93b605cb9b9071bf39a287020354251ade":{"kind":"REWORK","_number":1,"created":"2021-03-15 11:06:42.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/46/780546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/46/780546/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/46/780546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/46/780546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/46/780546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/46/780546/1"}}},"commit":{"parents":[{"commit":"d24fd0acf3fa31ea559ba89df371a9379cbdb1b0","subject":"Remove usage of deprecated params for nova::network::neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d24fd0acf3fa31ea559ba89df371a9379cbdb1b0"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-03-15 11:05:23.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-03-15 11:05:25.000000000","tz":330},"subject":"Drop CentOS8 based jobs for Wallaby","message":"Drop CentOS8 based jobs for Wallaby\n\nRDO wallaby release will be CentOS8-Stream only,\nLet\u0027s drop CentOS8 based jobs and Run CentOS8-Stream\nbased jobs as voting.\n\nChange-Id: Ieda2cd0e7d3f25b99c28d5b338ea45f84b8a8a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2b19cc93b605cb9b9071bf39a287020354251ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2b19cc93b605cb9b9071bf39a287020354251ade"}]},"branch":"refs/heads/master"},"eabad465d8656c57a8af0f2612fc1f6cac618cfd":{"kind":"REWORK","_number":2,"created":"2021-04-06 06:46:49.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/46/780546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/46/780546/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/46/780546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/46/780546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/46/780546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/46/780546/2"}}},"commit":{"parents":[{"commit":"da8313ff09c03fb941981f5d118b368cd416cac9","subject":"Unpin non-openstack puppet-modules in master branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/da8313ff09c03fb941981f5d118b368cd416cac9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-04-06 06:42:47.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-04-06 06:45:15.000000000","tz":330},"subject":"Drop CentOS8 based jobs for Wallaby","message":"Drop CentOS8 based jobs for Wallaby\n\nRDO wallaby release will be CentOS8-Stream only,\nLet\u0027s drop CentOS8 based jobs and Run CentOS8-Stream\nbased jobs as voting.\n\nNot removing packstack-centos8(just switching it to use\nc8-stream) job as it\u0027s being used in python-tempestconf but\nit removal[1] is being blocked due to other issues, this job\ncan be cleaned later once [1] is merged.\n\n[1] https://review.opendev.org/c/osf/python-tempestconf/+/780792\n\nChange-Id: Ieda2cd0e7d3f25b99c28d5b338ea45f84b8a8a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eabad465d8656c57a8af0f2612fc1f6cac618cfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eabad465d8656c57a8af0f2612fc1f6cac618cfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
