)]}'
{"id":"x%2Fpackstack~784749","triplet_id":"x%2Fpackstack~stable%2Fwallaby~I79deca7c655161d46ecb9af4efd884d18549e753","project":"x/packstack","branch":"stable/wallaby","topic":"wallaby","hashtags":[],"change_id":"I79deca7c655161d46ecb9af4efd884d18549e753","subject":"Fix .gitreview for stable/wallaby","status":"MERGED","created":"2021-04-05 08:24:06.000000000","updated":"2021-04-06 07:24:16.000000000","submitted":"2021-04-06 07:24:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784749-wallaby","meta_rev_id":"fd64a4315c1a0ce5ed09ffe85aa7f303b5c29753","_number":784749,"virtual_id_number":784749,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 07:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2021-04-06 06:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2021-04-06 06:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-05 09:40:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 06:31:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"3492fa36a8de2f3e3cb48a31d421ecc56d790e36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-04-05 08:24:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1edd06669a0b273a4fbdc96210b37f2ef549b82b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 09:40:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee0776b3be94a8c9d36b60651b45e8a : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/054515f7400147d2a62fe4145fdcfacc : SUCCESS in 3m 38s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/74bfb581646646bfa0e209e896226bd2 : SUCCESS in 54m 37s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/207c911840e74c52b42b35ab50f08998 : SUCCESS in 41m 07s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/636937a2fb444cf290f014d46964035a : POST_FAILURE in 8m 49s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3671fa962b4d4df382a7cc57e48a287d : SUCCESS in 50m 00s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f05ec1cec26942399b68e803549e95d4 : SUCCESS in 54m 35s (non-voting)\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ecd6185b82284c2f9f31f8e56110208a : SUCCESS in 48m 07s (non-voting)\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d8bd0d19524c4b44a3c19a64bd9dde42 : SUCCESS in 47m 47s (non-voting)\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5c26ac8ef8ef43a3a2c281e2e652d2e2 : SUCCESS in 59m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f81b9f1cd298b1089a00b78ba08bde4c4625005c","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-04-05 10:00:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"13a39c7f08b4861ded105884ccfb610bcf0e4f53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 11:14:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c834b6dab354582b7b6d84cc3545773 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94317b4a7f4849c5a7ca9d2639c6007f : SUCCESS in 4m 22s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2024c2ccd90644c78f3b638284570653 : SUCCESS in 58m 11s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/dfff28ff845f43148ba59baaaa1f9884 : SUCCESS in 47m 59s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/008d4928527d4374b0362b83db2ce1f9 : SUCCESS in 57m 41s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7d75a66cb1634d27ac1b9a24a629d898 : SUCCESS in 57m 54s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8635471a3b8348ddb28082467909cdce : SUCCESS in 41m 58s (non-voting)\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/86ba99d32fdc4815931de1bc1e4a4a57 : SUCCESS in 45m 48s (non-voting)\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/cb3f074598bf46798ffdc2bb7f91fc48 : SUCCESS in 47m 22s (non-voting)\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1a7aabfa48a84984ac05f09247114c32 : SUCCESS in 58m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3855de147e65a6eee5096a55c4648128b0d68c8d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-04-06 06:31:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bdc9e6147acfe2df84720747a26dd87d12f9e2bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 06:31:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33ee54a5b4489b34d86c7b5e53f4794a807a7da6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 07:24:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10398eb6eb0545d8af010377e4fdae0f : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a612f64d929c4adb857c1c17e7306813 : SUCCESS in 6m 41s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b065ad470aaf4545b2245462d34f6940 : SUCCESS in 42m 00s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1addc79f663945ecbdad1f9b42d86d64 : SUCCESS in 46m 22s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/faf6adf9f8ac432d8a7252a95eac5fd2 : SUCCESS in 46m 00s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a9377459f373485687badf6b68a702b3 : SUCCESS in 45m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fd64a4315c1a0ce5ed09ffe85aa7f303b5c29753","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 07:24:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d8cc0fb21bc0650b511bcfc06450f6f4634fb78","revisions":{"0d8cc0fb21bc0650b511bcfc06450f6f4634fb78":{"kind":"REWORK","_number":1,"created":"2021-04-05 08:24:06.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/49/784749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/49/784749/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/49/784749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/49/784749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/49/784749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/49/784749/1"}}},"commit":{"parents":[{"commit":"da8313ff09c03fb941981f5d118b368cd416cac9","subject":"Unpin non-openstack puppet-modules in master branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/da8313ff09c03fb941981f5d118b368cd416cac9"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-04-05 08:23:41.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-04-05 08:24:07.000000000","tz":120},"subject":"Fix .gitreview for stable/wallaby","message":"Fix .gitreview for stable/wallaby\n\nChange-Id: I79deca7c655161d46ecb9af4efd884d18549e753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0d8cc0fb21bc0650b511bcfc06450f6f4634fb78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0d8cc0fb21bc0650b511bcfc06450f6f4634fb78"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
