)]}'
{"id":"x%2Fpackstack~788907","triplet_id":"x%2Fpackstack~master~I51902dfef6653f96555bb9a8c994064eb9b4ae78","project":"x/packstack","branch":"master","topic":"aodh-service_credentials","hashtags":[],"change_id":"I51902dfef6653f96555bb9a8c994064eb9b4ae78","subject":"Replace deprecated aodh::auth","status":"MERGED","created":"2021-04-30 09:00:53.000000000","updated":"2021-05-03 04:56:59.000000000","submitted":"2021-05-03 04:56:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788907-aodh-service_credentials","meta_rev_id":"450526dd10539b188918731e8de632a686abec7e","_number":788907,"virtual_id_number":788907,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-03 04:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2021-05-03 04:04:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-30 15:11:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2021-05-03 04:04:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 09:30:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-30 09:41:52.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"CC"},{"updated":"2021-04-30 15:11:08.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2021-04-30 15:11:08.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2021-04-30 15:11:08.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2021-05-03 04:04:18.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"e7e0086d3c15d468be49a4774dfaeafe528d1b60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-30 09:00:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e0cede04b84db8ffa77ae7e06efc1fd330ecf03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-30 09:01:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f43eb65bac3090a1bdb174b4e00482739b393606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 09:30:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ffe2e7e83014a58b40e4aded2ae2136 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/518132993fb94d35b95bf37a3de29375 : SUCCESS in 5m 03s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9b3777a333eb4d5bbeff4b2807ec9452 : POST_FAILURE in 23m 50s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/41a71278b08c4b6ea542931a755db96e : POST_FAILURE in 11m 36s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d299a7c7425848e4881113396962c9c6 : POST_FAILURE in 10m 00s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/45b1c18ded6640aab0e0fbd1e7110aac : POST_FAILURE in 13m 13s","accounts_in_message":[],"_revision_number":2},{"id":"85125d1beaffb704ac99c7b349922636bba388da","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-04-30 09:41:52.000000000","message":"Patch Set 2:\n\nhttps://review.opendev.org/c/x/packstack/+/788910 should fix the CI issue.","accounts_in_message":[],"_revision_number":2},{"id":"a99f1722d793dbf30d3aba7a948b195987dbf5c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-04-30 14:12:46.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fd8bb3f3bb64f92682e8e71daee7c448b3e8b73e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 15:06:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a8db650842a4f649d2716f77ed766b8 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f8614d07d8947f7a3ef296cc08e1553 : SUCCESS in 4m 13s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c63564fa228a4e7f82befb410df1b898 : SUCCESS in 43m 41s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3410533877024ae18b3733ca8ce3d193 : SUCCESS in 42m 14s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/07ce35c2738c4d80b5522027f6cc85ab : SUCCESS in 48m 27s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8aaaa442aaa14527b32a3dc2e8566007 : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":3},{"id":"6ca0d5caf43452fd40e904f029fb8e4a8ccb39cb","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-04-30 15:11:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5ebb2ba6d7ab52314fbaaeb44d50df6ec1b7d921","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-03 04:04:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6ef60880db7fbb8adf3de7be0790f13f5ce29d9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 04:04:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2e91216951700fbfad55652bc348d6a0e32eea5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 04:56:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c61ac0403825402b9cf83a37371376ef : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8024406d4c3443a883583c84c0d67061 : SUCCESS in 4m 04s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a88906c9ea5747bc84601649143bca27 : SUCCESS in 42m 06s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6c5df73b0ac24209ac7e8fb62f9ca1a8 : SUCCESS in 41m 37s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a50ebe9279dc480b914e4823e7955889 : SUCCESS in 51m 25s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2ec57012bec44702b0cd9f6aed1cc1c2 : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":3},{"id":"450526dd10539b188918731e8de632a686abec7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 04:56:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e1f1579d26ebb40f37991b0d2ff689fe28ae60d7","revisions":{"c5c50538936ce5483981904ef2d149ecc01d9ba7":{"kind":"REWORK","_number":1,"created":"2021-04-30 09:00:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/788907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/788907/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/788907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/788907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/788907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/788907/1"}}},"commit":{"parents":[{"commit":"eabad465d8656c57a8af0f2612fc1f6cac618cfd","subject":"Drop CentOS8 based jobs for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eabad465d8656c57a8af0f2612fc1f6cac618cfd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 09:00:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 09:00:51.000000000","tz":540},"subject":"Replace deprecated aodh::auth","message":"Replace deprecated aodh::auth\n\nThe aodh::auth class was deprecated in favor of the new\naodh::service_credentials class.\n\nChange-Id: I51902dfef6653f96555bb9a8c994064eb9b4ae78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c5c50538936ce5483981904ef2d149ecc01d9ba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c5c50538936ce5483981904ef2d149ecc01d9ba7"}]},"branch":"refs/heads/master"},"949fd7db44c7e7fa71a82ac9eb92bca26f8b0d1b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-30 09:01:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/788907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/788907/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/788907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/788907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/788907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/788907/2"}}},"commit":{"parents":[{"commit":"eabad465d8656c57a8af0f2612fc1f6cac618cfd","subject":"Drop CentOS8 based jobs for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eabad465d8656c57a8af0f2612fc1f6cac618cfd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 09:00:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 09:01:31.000000000","tz":540},"subject":"Replace deprecated aodh::auth","message":"Replace deprecated aodh::auth\n\nThe aodh::auth class was deprecated in favor of the new\naodh::service_credentials class.\n\nDepends-on: https://review.opendev.org/772694\nChange-Id: I51902dfef6653f96555bb9a8c994064eb9b4ae78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/949fd7db44c7e7fa71a82ac9eb92bca26f8b0d1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/949fd7db44c7e7fa71a82ac9eb92bca26f8b0d1b"}]},"branch":"refs/heads/master"},"e1f1579d26ebb40f37991b0d2ff689fe28ae60d7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-30 14:12:46.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/07/788907/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/788907/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/788907/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/788907/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/788907/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/788907/3"}}},"commit":{"parents":[{"commit":"e11676c1253167431d860200492bc71d0ea21f89","subject":"Fix CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e11676c1253167431d860200492bc71d0ea21f89"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 09:00:12.000000000","tz":540},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2021-04-30 14:12:46.000000000","tz":0},"subject":"Replace deprecated aodh::auth","message":"Replace deprecated aodh::auth\n\nThe aodh::auth class was deprecated in favor of the new\naodh::service_credentials class.\n\nDepends-on: https://review.opendev.org/772694\nChange-Id: I51902dfef6653f96555bb9a8c994064eb9b4ae78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e1f1579d26ebb40f37991b0d2ff689fe28ae60d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e1f1579d26ebb40f37991b0d2ff689fe28ae60d7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
