)]}'
{"id":"x%2Fpackstack~788910","triplet_id":"x%2Fpackstack~master~If98beab4946f0a97f3308db050318000f144ba07","project":"x/packstack","branch":"master","hashtags":[],"change_id":"If98beab4946f0a97f3308db050318000f144ba07","subject":"Fix CI jobs","status":"MERGED","created":"2021-04-30 09:40:52.000000000","updated":"2021-04-30 14:10:03.000000000","submitted":"2021-04-30 14:10:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788910","meta_rev_id":"05fbee4fd2cab92c2f1fba3ab3d97b4b77eb36ab","_number":788910,"virtual_id_number":788910,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-30 14:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2021-04-30 13:13:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2021-04-30 13:13:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 09:41:26.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-04-30 09:41:26.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-04-30 09:56:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-30 12:40:20.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"4dbc12fd3f61a88391cd6474a82a1b3ad53c617c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-04-30 09:40:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40e45a2f9d98ae2bcfc68daf847302c58eb9be6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 09:56:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/197b646e508d4ab29f653c2f13751ad5 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84a1998f81614155ad5737c491b1de51 : SUCCESS in 4m 33s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/144e2e59e5624e04b2a8852a3092bb62 : FAILURE in 11m 04s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/06d9e1780a664323981e34ff711e3f16 : FAILURE in 10m 45s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/4e260873410745cca96e18d8efe71d69 : FAILURE in 10m 15s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b590644f0d1044d8b07bcf68e679111e : FAILURE in 11m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8fa38432abbaba931e4209ee37897c864ab430ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-04-30 10:00:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a507153427e0c3edac2a5e21e21b708909d9763","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-04-30 10:28:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9f09944cab133f6cd770d77a7cec40dc4d4e42c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 11:34:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2395c364c3c440c896b055786964c5ad : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd5e10c900e64415bc089eaac2275e6d : SUCCESS in 3m 59s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e7fd32e6ee1947af82c789b70c9f0d32 : SUCCESS in 51m 03s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d0db5997ccf04e3296817e799fa7d4f5 : SUCCESS in 53m 52s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d47b6751b85a436da3bbd6af808bd057 : SUCCESS in 59m 51s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e93e983cc8704673a8baf1e51c005af0 : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":3},{"id":"bc11e1006cca8f29f15c4776a8c679f863c0dbd9","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-04-30 13:13:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9fecd78a1dce6342029bb21efd09ff1f0ec812cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 13:14:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a66dbfb8ee9c2d58a9838ea5ec9c4b10527eae93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 14:09:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e2edd359f14eeb808cf0e2dcd2dc5a : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90f682718b23421d9a005bfb4940b126 : SUCCESS in 4m 14s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c9732e8da10243af8ffbb5bd7edc4bc1 : SUCCESS in 48m 56s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/74d94c6a87524b7487bf368d9a070837 : SUCCESS in 44m 36s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6f4ac19518b240e1b4a029385198413d : SUCCESS in 50m 17s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8d4e73319a7345e999559075804b3038 : SUCCESS in 50m 39s","accounts_in_message":[],"_revision_number":3},{"id":"05fbee4fd2cab92c2f1fba3ab3d97b4b77eb36ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 14:10:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e11676c1253167431d860200492bc71d0ea21f89","revisions":{"9e9bbe84900ae0c05b343d2e33d1ccc12cf642c1":{"kind":"REWORK","_number":1,"created":"2021-04-30 09:40:52.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/10/788910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/10/788910/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/10/788910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/10/788910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/10/788910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/10/788910/1"}}},"commit":{"parents":[{"commit":"eabad465d8656c57a8af0f2612fc1f6cac618cfd","subject":"Drop CentOS8 based jobs for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eabad465d8656c57a8af0f2612fc1f6cac618cfd"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-04-30 09:39:56.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-04-30 09:39:56.000000000","tz":120},"subject":"Fix branch for puppetlabs Puppet modules","message":"Fix branch for puppetlabs Puppet modules\n\nThe branch for most Puppet modules in the github.com/puppetlabs\nnamespace has changed from \u0027master\u0027 to \u0027main\u0027.\n\nChange-Id: If98beab4946f0a97f3308db050318000f144ba07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9e9bbe84900ae0c05b343d2e33d1ccc12cf642c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9e9bbe84900ae0c05b343d2e33d1ccc12cf642c1"}]},"branch":"refs/heads/master"},"55cb570794911bf4be68ac9d3228d2cba9eaa738":{"kind":"REWORK","_number":2,"created":"2021-04-30 10:00:16.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/10/788910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/10/788910/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/10/788910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/10/788910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/10/788910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/10/788910/2"}}},"commit":{"parents":[{"commit":"eabad465d8656c57a8af0f2612fc1f6cac618cfd","subject":"Drop CentOS8 based jobs for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eabad465d8656c57a8af0f2612fc1f6cac618cfd"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-04-30 09:39:56.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-04-30 09:58:57.000000000","tz":120},"subject":"Fix CI jobs","message":"Fix CI jobs\n\nThe branch for most Puppet modules in the github.com/puppetlabs\nnamespace has changed from \u0027master\u0027 to \u0027main\u0027.\n\nAlso, the scheduler_default_filters parameter was removed from\nclass nova::scheduler::filter in puppet-nova by [1].\n\n[1] - https://review.opendev.org/c/openstack/puppet-nova/+/787967\n\nChange-Id: If98beab4946f0a97f3308db050318000f144ba07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/55cb570794911bf4be68ac9d3228d2cba9eaa738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/55cb570794911bf4be68ac9d3228d2cba9eaa738"}]},"branch":"refs/heads/master"},"e11676c1253167431d860200492bc71d0ea21f89":{"kind":"REWORK","_number":3,"created":"2021-04-30 10:28:08.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/10/788910/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/10/788910/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/10/788910/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/10/788910/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/10/788910/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/10/788910/3"}}},"commit":{"parents":[{"commit":"eabad465d8656c57a8af0f2612fc1f6cac618cfd","subject":"Drop CentOS8 based jobs for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eabad465d8656c57a8af0f2612fc1f6cac618cfd"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-04-30 09:39:56.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-04-30 10:27:18.000000000","tz":120},"subject":"Fix CI jobs","message":"Fix CI jobs\n\nThe branch for most Puppet modules in the github.com/puppetlabs\nnamespace has changed from \u0027master\u0027 to \u0027main\u0027.\n\nAlso, the scheduler_default_filters parameter was removed from\nclass nova::scheduler::filter in puppet-nova by [1].\n\nAnd set ssl_verify_client parameter to apache::vhost to \u0027optional\u0027,\nsince using a boolean \u0027false\u0027 is no longer valid [2].\n\n[1] - https://review.opendev.org/c/openstack/puppet-nova/+/787967\n[2] - https://github.com/puppetlabs/puppetlabs-apache/blob/main/manifests/vhost.pp#L1769\n\nChange-Id: If98beab4946f0a97f3308db050318000f144ba07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e11676c1253167431d860200492bc71d0ea21f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e11676c1253167431d860200492bc71d0ea21f89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
