)]}'
{"id":"x%2Fpackstack~789379","triplet_id":"x%2Fpackstack~master~Icf925c479a9fc05fe71d8315306cb390fd7a57d9","project":"x/packstack","branch":"master","topic":"abs-name","hashtags":[],"change_id":"Icf925c479a9fc05fe71d8315306cb390fd7a57d9","subject":"Replace absolute class names","status":"MERGED","created":"2021-05-03 16:08:58.000000000","updated":"2021-05-04 15:08:52.000000000","submitted":"2021-05-04 15:08:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":9,"has_review_started":true,"submission_id":"789379-abs-name","meta_rev_id":"8b9c992e9c64b29dd71cfc408a8d156d43eda8dc","_number":789379,"virtual_id_number":789379,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-04 15:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2021-05-04 12:15:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-04 13:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-04 13:44:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-03 17:12:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-04 07:35:57.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2021-05-04 08:07:25.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"CC"},{"updated":"2021-05-04 10:38:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2021-05-04 12:15:15.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-05-04 13:44:13.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"}],"messages":[{"id":"62d023516f3f80783be4e85ddd2410d6896f4db4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-03 16:08:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"595f7ba54ad1244a2775f9a27da9aee7be54cb12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 17:12:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/491e77a7974841d7bd67f5425977ae42 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/340479d9eb1949e28ba2f2619f5026d7 : SUCCESS in 5m 24s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/dc3966d696ed4f248d0a786531ac5d8e : SUCCESS in 38m 12s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b3ced357393c4682bfc55fa7c91a4704 : POST_FAILURE in 9m 13s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/254a647a064d49be80848268868cb705 : SUCCESS in 57m 20s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/70e4492184d04c51a429fe619c29c026 : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":1},{"id":"70423c3955eeb406b14c03b2522828b43395c38a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-05-04 07:35:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a250c6c2f5dd01a641302cbe815d7f0bfa5eef6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-04 07:47:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d3d4f4e0fdc75fba15a3c34f8f3615045bf21fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-04 07:47:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"33a970ea3c20a4196e5a727f97a92a176d6dc837","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-05-04 08:07:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3cd4726389102a6d8313eaf4ec0f07a6670897f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 08:45:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3622fac7a0d498caec726c09490c9d7 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54047f7ffb9d454798a93ae2942e5d19 : SUCCESS in 3m 56s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6c7194224ee24642910c95e2e3e118ac : SUCCESS in 33m 13s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f07ed1635e8a4048bfe85ea3af7d86ec : SUCCESS in 40m 27s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b896bd1187d94240a45f94997772df9e : SUCCESS in 56m 24s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8f207dbb59bb40f9bdb0f725374734c4 : SUCCESS in 44m 03s","accounts_in_message":[],"_revision_number":2},{"id":"160f56fdf57aa9531b88c0d5b81adff6ea790281","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-04 10:38:21.000000000","message":"Patch Set 2:\n\n(8 comments)\n\nsome extra spaces injected in include, may be the script used for replacement injected it, would be good to clean those extra space for clarity.","accounts_in_message":[],"_revision_number":2},{"id":"ebc246712914a4147372bed398be06d6964ed8cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-04 11:22:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5889dc865c0a3f188115a16b67cdcc1dd6855325","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-04 11:24:21.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e (8 comments)\n\u003e \n\u003e some extra spaces injected in include, may be the script used for replacement injected it, would be good to clean those extra space for clarity.\n\nThanks for the good catch. Indeed I accidentally injected one space by my sed command.\nThe updated version should have all extra spaces removed.","accounts_in_message":[],"_revision_number":2},{"id":"e7d1366d76aca34f47e82330296ac48210ccf970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 12:10:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6521abc03cda41b2bcbd10edc36eb574 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8746987124c24ec5a8edeb2947eac3c7 : SUCCESS in 3m 48s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1f805717214648de92b4fbf799066c63 : SUCCESS in 43m 55s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/9963d2f277a94432b604d9a73d1132c5 : SUCCESS in 41m 37s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b84ab6e002374a37b6fe6f1666a57efd : SUCCESS in 46m 41s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8d107b6a4e2b468689c2a32ac0b1426e : SUCCESS in 46m 28s","accounts_in_message":[],"_revision_number":3},{"id":"7d7b04521afe08aa9a4c773a9285e428f7884246","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-04 12:15:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"60f11a3fd8b72786a25c597c94c76a43fda2e6fc","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-05-04 13:44:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"939b87ae18f92159f781c5befa25c3df9f9d7850","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 13:44:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"44a7383f9dd4f443c3480dd1983ce4f3725858b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 15:08:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c596f06664d40cf86f4dcd2c2721a69 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/471bef6048284a3ab099ba36d25bdc29 : SUCCESS in 4m 15s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e610146ee68146ae9171bd84042a7c1f : SUCCESS in 43m 07s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d70d8fc173f74b218bec4671af5bf643 : SUCCESS in 1h 09m 24s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7c41e4c0843845c98b79b6533cc8ab1c : SUCCESS in 47m 55s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c5d188066b4a431babb1bf32598612cd : SUCCESS in 1h 09m 15s","accounts_in_message":[],"_revision_number":3},{"id":"8b9c992e9c64b29dd71cfc408a8d156d43eda8dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 15:08:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"381feb7e090f9ab6a9ab00c6eb2402730a653c6b","revisions":{"84c0ce6932bac9e833c375171c23a7fc927e33a3":{"kind":"REWORK","_number":1,"created":"2021-05-03 16:08:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/789379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/79/789379/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/79/789379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/79/789379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/79/789379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/79/789379/1"}}},"commit":{"parents":[{"commit":"e68e0e8232f2502c56df758cb7418063ff9906b9","subject":"Merge \"Replace deprecated ceilometer::agent::auth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e68e0e8232f2502c56df758cb7418063ff9906b9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-03 16:07:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-03 16:07:02.000000000","tz":540},"subject":"Replace absolute class names","message":"Replace absolute class names\n\nSince Puppet 4 all class names are handled as absolute class names\nand we don\u0027t neet to explicitly use $::{name}.\n\nChange-Id: Icf925c479a9fc05fe71d8315306cb390fd7a57d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/84c0ce6932bac9e833c375171c23a7fc927e33a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/84c0ce6932bac9e833c375171c23a7fc927e33a3"}]},"branch":"refs/heads/master"},"e04a4923526c3a2d5ba4974a9a2fe417921b047a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-04 07:47:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/789379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/79/789379/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/79/789379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/79/789379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/79/789379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/79/789379/2"}}},"commit":{"parents":[{"commit":"e68e0e8232f2502c56df758cb7418063ff9906b9","subject":"Merge \"Replace deprecated ceilometer::agent::auth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e68e0e8232f2502c56df758cb7418063ff9906b9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-03 16:07:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 07:47:30.000000000","tz":0},"subject":"Replace absolute class names","message":"Replace absolute class names\n\nSince Puppet 4 all class names are handled as absolute class names\nand we don\u0027t need to explicitly use $::{name}.\n\nChange-Id: Icf925c479a9fc05fe71d8315306cb390fd7a57d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e04a4923526c3a2d5ba4974a9a2fe417921b047a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e04a4923526c3a2d5ba4974a9a2fe417921b047a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"381feb7e090f9ab6a9ab00c6eb2402730a653c6b":{"kind":"REWORK","_number":3,"created":"2021-05-04 11:22:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/789379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/79/789379/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/79/789379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/79/789379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/79/789379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/79/789379/3"}}},"commit":{"parents":[{"commit":"e68e0e8232f2502c56df758cb7418063ff9906b9","subject":"Merge \"Replace deprecated ceilometer::agent::auth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e68e0e8232f2502c56df758cb7418063ff9906b9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-03 16:07:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 11:22:35.000000000","tz":540},"subject":"Replace absolute class names","message":"Replace absolute class names\n\nSince Puppet 4 all class names are handled as absolute class names\nand we don\u0027t neet to explicitly use $::{name}.\n\nChange-Id: Icf925c479a9fc05fe71d8315306cb390fd7a57d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/381feb7e090f9ab6a9ab00c6eb2402730a653c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/381feb7e090f9ab6a9ab00c6eb2402730a653c6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
