)]}'
{"id":"x%2Fpackstack~789801","triplet_id":"x%2Fpackstack~stable%2Fwallaby~Id721cd84740adb9d459c1a5ad5cba919735b6f01","project":"x/packstack","branch":"stable/wallaby","hashtags":[],"change_id":"Id721cd84740adb9d459c1a5ad5cba919735b6f01","subject":"Use ensure_packages to install python3-redis","status":"MERGED","created":"2021-05-05 11:46:51.000000000","updated":"2021-05-07 08:45:08.000000000","submitted":"2021-05-07 08:45:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789801","cherry_pick_of_change":789735,"cherry_pick_of_patch_set":1,"meta_rev_id":"a1aa9ee3ada0773b264e9ca59a84cc0e6736483b","_number":789801,"virtual_id_number":789801,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-07 08:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2021-05-07 07:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-06 08:06:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2021-05-07 07:45:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 11:46:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2021-05-05 12:10:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 08:06:48.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"8ed9074456d863523fc2b149804bb746b9f45348","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-05 11:46:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"65df9d4922845500ce889bb3fdea5342d2d17167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:10:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13e054c58c1f4d2686cd11d2ff71697e : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3f3f9de30fa425eb399a9a67b455b18 : SUCCESS in 4m 50s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/71c9df30505640d09f9d13c0b8346235 : POST_FAILURE in 19m 02s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/48dbbae9daac456db5ad3f3921a97450 : POST_FAILURE in 19m 06s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a06ad9ad6013477cb7a2835e297c3785 : POST_FAILURE in 10m 02s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5a61be2df4c64df1a29c5c35d689ef1a : POST_FAILURE in 17m 56s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9a52d92dae43486aba00a9a763aef937 : POST_FAILURE in 10m 06s (non-voting)\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bcf914cb497e418fb0463c786648e252 : POST_FAILURE in 9m 58s (non-voting)\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6cb36d7a1b34471c844c00b58de27e4c : POST_FAILURE in 8m 58s (non-voting)\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6c1831112c28477b9e8853f3860c89e3 : POST_FAILURE in 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f97efba99f0bd392c3e5f54475aa8f1a07641c2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-05-05 12:49:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"798c5b665f1e18a08c555c22660dbc1898d9ad71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 13:25:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e546e8b3cdc9437c9b43686045ee9c42 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67b0e9ce07054ffa8a7b11106dae8cb3 : SUCCESS in 4m 09s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/dc8c389850a147918a14c9721c328a69 : FAILURE in 12m 10s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/213912ce41664d0b948a41ee6a1d9cb4 : FAILURE in 15m 38s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d209ebc11de04862bda17932ed7185e5 : FAILURE in 11m 32s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/36fb112c741944d3823138cbf598bab8 : FAILURE in 12m 46s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/cebf8a686d6245f599d5931de76d8727 : FAILURE in 11m 23s (non-voting)\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c45806d97edf4245b1660421864f10b5 : FAILURE in 17m 39s (non-voting)\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b93c7a018da1448a916787444531391c : FAILURE in 16m 22s (non-voting)\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/24b4acaa4e7a44f78a3fa2948bb24308 : FAILURE in 18m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"37b686b55f22cb2a609d3c1fbfee42c94ac8a730","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-05-05 13:35:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c30f09bce2ea9957d07bf3ba5467f12564fdb399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 14:37:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/730dd8dc67dd44dfbd821e1ae55c1982 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d896bddb3a34c0d9e54a2eb56c7fa1a : SUCCESS in 4m 02s\n- packstack-centos8-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8dbccf83be2e4d6ebb5b1864677779b1 : FAILURE in 24m 09s\n- packstack-centos8-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bef8a8c18aa2439d8bb8ed7e23a154c7 : FAILURE in 30m 07s\n- packstack-centos8-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e89fac4372d54c73ada15ef2f6e8fae7 : FAILURE in 22m 42s\n- packstack-centos8-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1d1c00fd7bc1409590cc44598b2f0ae0 : FAILURE in 38m 56s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0757ae6aa3b94fc1b4297acdc8f20728 : SUCCESS in 48m 28s (non-voting)\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/915ad6633aeb4916bbe4f35786eae864 : SUCCESS in 48m 14s (non-voting)\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b277183b75d24268bb6a924b30ca48cb : SUCCESS in 47m 00s (non-voting)\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2ab6121c4fd94d06b58ddd53c93a9c47 : SUCCESS in 57m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c26d8f0b17909252f2c00c9451e93aad3df347a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-05-05 15:03:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fe7ba3b91745ce5ede04b241df331bd39e29c9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 17:07:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9515be79ad64f40ab3fb6922460e40b : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f701df7013246e98084f69688e273aa : SUCCESS in 4m 47s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/af1d5678d4c340c59c058ab884c5326f : TIMED_OUT in 2h 01m 52s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/114db9b34f344289bb13dadb83b75c81 : SUCCESS in 42m 05s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8652dc8c648e4357a710ec0a64d8ac7f : SUCCESS in 1h 04m 53s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/dfb75740f83b43af97715c17595fabf1 : SUCCESS in 1h 10m 01s","accounts_in_message":[],"_revision_number":4},{"id":"3b9c3a4cbc985cf081fa4e1a8f4415d25b397b9e","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-05-06 07:02:41.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9734e4f4e135fc07027198597685a67e74abc329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 08:01:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95f1692199f44872b4938116b86cc2c3 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f131b640160f49f09e971503af30a704 : SUCCESS in 4m 41s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/179bc4e17dcd4ade889ed72497221d0b : SUCCESS in 49m 32s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0de1a283b62a4a089652926b0c672611 : SUCCESS in 51m 03s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/be05cb9afb374971bad40f9919fee6be : SUCCESS in 54m 39s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1359ab0538834b6baaa0dfe91836669a : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":4},{"id":"4333abbf173981a6e283b0853a14db5d166b7545","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-05-06 08:06:52.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9f4a2c2575ee809a40c78c32ca364cc92d50383d","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-05-07 07:45:11.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d37f502525153882ca9683d26300c6b46c93be9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 07:45:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"af33b44beaacd1c95adfaab945d1c2fdd141bd5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 08:45:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f37d060cb34674b47a00c8d63ecf3b : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0043795a73b0497884f42b2afe5e9c3d : SUCCESS in 4m 19s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/bf15fb426b6e40128e702b219ac39660 : SUCCESS in 50m 19s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d8a6912ec3c94edf802a245cce84a38f : SUCCESS in 47m 47s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/82fa1bb9e785417d8f3ac82fade3802e : SUCCESS in 45m 07s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/618f0171b6ae48f69985cc7e62592227 : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":4},{"id":"a1aa9ee3ada0773b264e9ca59a84cc0e6736483b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 08:45:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d39d4e3ca70261c07996e5d29a7648a7cada9053","revisions":{"ff1d17637362593d3926d9f09b8e124587008205":{"kind":"REWORK","_number":1,"created":"2021-05-05 11:46:51.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/01/789801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/01/789801/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/01/789801/1"}}},"commit":{"parents":[{"commit":"0d8cc0fb21bc0650b511bcfc06450f6f4634fb78","subject":"Fix .gitreview for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0d8cc0fb21bc0650b511bcfc06450f6f4634fb78"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-05-05 08:54:11.000000000","tz":120},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2021-05-05 11:46:51.000000000","tz":0},"subject":"Use ensure_packages to install python3-redis","message":"Use ensure_packages to install python3-redis\n\nIn the ceilometer manifest, we were using ensure_resource to make\nsure the python3-redis package was installed when needed. However,\npuppet-aodh changed to use ensure_packages with [1], which breaks\nus. Moving to use ensure_packages as well.\n\n[1] - https://review.opendev.org/c/openstack/puppet-aodh/+/789419\n\nChange-Id: Id721cd84740adb9d459c1a5ad5cba919735b6f01\n(cherry picked from commit 33517d4b22cfbd2fad1ca57d4d57ac4a4101c5d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ff1d17637362593d3926d9f09b8e124587008205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ff1d17637362593d3926d9f09b8e124587008205"}]},"branch":"refs/heads/stable/wallaby"},"687e1c2d1f1b68b74b5310d18a39dae65ac4789a":{"kind":"REWORK","_number":2,"created":"2021-05-05 12:49:19.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/01/789801/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/01/789801/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/01/789801/2"}}},"commit":{"parents":[{"commit":"0d8cc0fb21bc0650b511bcfc06450f6f4634fb78","subject":"Fix .gitreview for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0d8cc0fb21bc0650b511bcfc06450f6f4634fb78"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-05-05 08:54:11.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-05-05 12:48:42.000000000","tz":120},"subject":"Use ensure_packages to install python3-redis","message":"Use ensure_packages to install python3-redis\n\nIn the ceilometer manifest, we were using ensure_resource to make\nsure the python3-redis package was installed when needed. However,\npuppet-aodh changed to use ensure_packages with [1], which breaks\nus. Moving to use ensure_packages as well.\n\nAlso, do a partial cherry-pick of [2] to fix CI jobs.\n\n[1] - https://review.opendev.org/c/openstack/puppet-aodh/+/789419\n[2] - https://review.opendev.org/c/x/packstack/+/788910\n\nChange-Id: Id721cd84740adb9d459c1a5ad5cba919735b6f01\n(cherry picked from commit 33517d4b22cfbd2fad1ca57d4d57ac4a4101c5d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/687e1c2d1f1b68b74b5310d18a39dae65ac4789a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/687e1c2d1f1b68b74b5310d18a39dae65ac4789a"}]},"branch":"refs/heads/stable/wallaby"},"311ab4e6e020d9e55a41c1e76cdb42b33e14d9eb":{"kind":"REWORK","_number":3,"created":"2021-05-05 13:35:57.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/01/789801/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/01/789801/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/01/789801/3"}}},"commit":{"parents":[{"commit":"0d8cc0fb21bc0650b511bcfc06450f6f4634fb78","subject":"Fix .gitreview for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0d8cc0fb21bc0650b511bcfc06450f6f4634fb78"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-05-05 08:54:11.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-05-05 13:35:37.000000000","tz":120},"subject":"Use ensure_packages to install python3-redis","message":"Use ensure_packages to install python3-redis\n\nIn the ceilometer manifest, we were using ensure_resource to make\nsure the python3-redis package was installed when needed. However,\npuppet-aodh changed to use ensure_packages with [1], which breaks\nus. Moving to use ensure_packages as well.\n\nAlso, do a partial cherry-pick of [2] to fix CI jobs, and set branch\nfor OpenStack Puppet Modules to stable/wallaby.\n\n[1] - https://review.opendev.org/c/openstack/puppet-aodh/+/789419\n[2] - https://review.opendev.org/c/x/packstack/+/788910\n\nChange-Id: Id721cd84740adb9d459c1a5ad5cba919735b6f01\n(cherry picked from commit 33517d4b22cfbd2fad1ca57d4d57ac4a4101c5d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/311ab4e6e020d9e55a41c1e76cdb42b33e14d9eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/311ab4e6e020d9e55a41c1e76cdb42b33e14d9eb"}]},"branch":"refs/heads/stable/wallaby"},"d39d4e3ca70261c07996e5d29a7648a7cada9053":{"kind":"REWORK","_number":4,"created":"2021-05-05 15:03:05.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/01/789801/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/01/789801/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/01/789801/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/01/789801/4"}}},"commit":{"parents":[{"commit":"0d8cc0fb21bc0650b511bcfc06450f6f4634fb78","subject":"Fix .gitreview for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0d8cc0fb21bc0650b511bcfc06450f6f4634fb78"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-05-05 08:54:11.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-05-05 15:02:19.000000000","tz":120},"subject":"Use ensure_packages to install python3-redis","message":"Use ensure_packages to install python3-redis\n\nIn the ceilometer manifest, we were using ensure_resource to make\nsure the python3-redis package was installed when needed. However,\npuppet-aodh changed to use ensure_packages with [1], which breaks\nus. Moving to use ensure_packages as well.\n\nAlso, do a partial cherry-pick of [2] to fix CI jobs, and set branch\nfor OpenStack Puppet Modules to stable/wallaby.\n\nAlso, cherry-pick [3] and set RDO Trunk repos to use centos8-wallaby.\n\n[1] - https://review.opendev.org/c/openstack/puppet-aodh/+/789419\n[2] - https://review.opendev.org/c/x/packstack/+/788910\n[3] - https://review.opendev.org/c/x/packstack/+/780546\n\nChange-Id: Id721cd84740adb9d459c1a5ad5cba919735b6f01\n(cherry picked from commit 33517d4b22cfbd2fad1ca57d4d57ac4a4101c5d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d39d4e3ca70261c07996e5d29a7648a7cada9053"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d39d4e3ca70261c07996e5d29a7648a7cada9053"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
