)]}'
{"id":"x%2Fpackstack~792370","triplet_id":"x%2Fpackstack~master~I744de28a25739c00f585b7d9c12627ce0ed902f2","project":"x/packstack","branch":"master","topic":"c9","hashtags":[],"change_id":"I744de28a25739c00f585b7d9c12627ce0ed902f2","subject":"Add support for EL9 and python3.9","status":"MERGED","created":"2021-05-20 12:54:07.000000000","updated":"2021-05-21 13:18:29.000000000","submitted":"2021-05-21 13:18:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792370-c9","meta_rev_id":"f7f885416ce4efa62246f7045ef007d61b5a2b4b","_number":792370,"virtual_id_number":792370,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-21 13:18:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2021-05-21 12:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-21 10:10:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2021-05-21 12:21:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"CC":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-20 13:35:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 19:35:02.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"CC"},{"updated":"2021-05-21 10:04:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2021-05-21 10:04:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-05-21 10:04:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"}],"messages":[{"id":"424b5b4b747e5bd8c8369d46daf5576dcbd78eae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-20 12:54:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6736dd92192f410c3d3a8513d6319c6a79717b33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 13:35:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/320d6203272e4150bbf8b825bb320e65 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b4c0e598a2440ee8bceb4ea2a5260f7 : FAILURE in 4m 23s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2d99bd919e844141b6cdee95a70276e0 : FAILURE in 39m 02s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/952fc02844ef48218e23bc697294717e : FAILURE in 28m 20s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e7d2573636ac4ed5a23ae598bda2737c : FAILURE in 37m 31s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/87e724c417094e1599461e235fd87051 : FAILURE in 36m 19s","accounts_in_message":[],"_revision_number":1},{"id":"058e97dd1f16807ba6ae41ce6dc9a6e7e4cfac91","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-05-20 19:35:02.000000000","message":"Patch Set 1:\n\ncould we add scen000 https://github.com/openstack/puppet-openstack-integration#description ?","accounts_in_message":[],"_revision_number":1},{"id":"3a5924b419b9ee12313be3e8ef352c8f5a04e94a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-21 05:40:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ebeded070b03767ca84d3be9c91c92c77ee66f42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-21 05:41:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8c0e2db21fa05284e0bdd75878727aa601eec634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 06:30:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b93c4d7923b44e789a7aefb2961f04b5 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f371492d2e34d398d3368ed8c504273 : SUCCESS in 4m 11s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/480fd45416314db180cd177d058dff83 : SUCCESS in 43m 53s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2545c1d8566245599350e0f460a1b52b : SUCCESS in 42m 01s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/28d7dcd73b994209b761b088c977831e : SUCCESS in 49m 08s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d16ca10488fe477784a044b734ae24a3 : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":3},{"id":"db19aa4bdf75119b852c2b16404df567b1bd553a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-21 06:39:20.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 1:\n\u003e \n\u003e could we add scen000 https://github.com/openstack/puppet-openstack-integration#description ?\n\nYes, can be added as followup if that\u0027s needed and useful.","accounts_in_message":[],"_revision_number":3},{"id":"19b38ee235305a2db970fa1609d6415c107330f1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-21 10:06:50.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e could we add scen000 https://github.com/openstack/puppet-openstack-integration#description ?\n\u003e \n\u003e Yes, can be added as followup if that\u0027s needed and useful.\n\nbtw scenario named 000 is already present but deploys all default services https://github.com/redhat-openstack/packstack/blob/master/tests/scenario000.sh","accounts_in_message":[],"_revision_number":3},{"id":"f0aa41c692b1432bd46dcc8985b1358fd503ff71","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-05-21 10:10:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ac6e08f0db7f0f33321f65ed4e8b0f93b25680b4","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-05-21 12:21:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9008da3f486597f8223b0b60fdc2168b0a07cf73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:22:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"37bcfb6c29c040024fa2c883e9b84cc1c1a17b08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 13:18:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bdcdb4acbe14c05996b034694d74f06 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc6f293951bc46e4891286376b148a6d : SUCCESS in 4m 26s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/158c2a82ab8f4ab69d0ab21124d22caf : SUCCESS in 36m 22s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/dc4e229ffad14429b2632e4038a00c4d : SUCCESS in 45m 02s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9046f2af5fe54dfda6c18795e5cb6a38 : SUCCESS in 49m 17s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/99e6811d96dd457690def4dcbd7b8d99 : SUCCESS in 49m 33s","accounts_in_message":[],"_revision_number":3},{"id":"f7f885416ce4efa62246f7045ef007d61b5a2b4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 13:18:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e11f86aecadd0a52eda4fbcdedd2ecfe6b5892b5","revisions":{"4c69fb8cdf6028ee6fea4a041f0b721b6fa4fc26":{"kind":"REWORK","_number":1,"created":"2021-05-20 12:54:07.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/70/792370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/70/792370/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/70/792370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/70/792370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/70/792370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/70/792370/1"}}},"commit":{"parents":[{"commit":"895b6e9e11b8ab437f595288ac1c8988348125ea","subject":"Use env from /usr/bin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/895b6e9e11b8ab437f595288ac1c8988348125ea"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-20 12:39:36.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-20 12:50:58.000000000","tz":330},"subject":"Add support for EL9 and python3.9","message":"Add support for EL9 and python3.9\n\nFollowing is done:-\n- Using distro module to get distribution as platform\n  module no longer provide that with python3.8+[1].\n- Use mariadb-server-galera as mariadb don\u0027t provide\n  mariadb-galera-server in EL9[2].\n- Ensure \u0027cronie\u0027 package is installed before attempting\n  start of crond service.\n\n[1] https://bugs.python.org/issue28167\n[2] https://src.fedoraproject.org/rpms/mariadb/c/b1bc71c1a\n\nChange-Id: I744de28a25739c00f585b7d9c12627ce0ed902f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4c69fb8cdf6028ee6fea4a041f0b721b6fa4fc26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4c69fb8cdf6028ee6fea4a041f0b721b6fa4fc26"}]},"branch":"refs/heads/master"},"76f471b56cf7430516237c57a786fe3ce13fdbc0":{"kind":"REWORK","_number":2,"created":"2021-05-21 05:40:33.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/70/792370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/70/792370/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/70/792370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/70/792370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/70/792370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/70/792370/2"}}},"commit":{"parents":[{"commit":"895b6e9e11b8ab437f595288ac1c8988348125ea","subject":"Use env from /usr/bin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/895b6e9e11b8ab437f595288ac1c8988348125ea"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-20 12:39:36.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-21 05:40:16.000000000","tz":330},"subject":"Add support for EL9 and python3.9","message":"Add support for EL9 and python3.9\n\nFollowing is done:-\n- Using distro module to get distribution as platform\n  module no longer provide that with python3.8+[1].\n- Use mariadb-server-galera as mariadb don\u0027t provide\n  mariadb-galera-server in EL9[2].\n- Ensure \u0027cronie\u0027 package is installed before attempting\n  start of crond service.\n\n[1] https://bugs.python.org/issue28167\n[2] https://src.fedoraproject.org/rpms/mariadb/c/b1bc71c1a\n\nChange-Id: I744de28a25739c00f585b7d9c12627ce0ed902f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/76f471b56cf7430516237c57a786fe3ce13fdbc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/76f471b56cf7430516237c57a786fe3ce13fdbc0"}]},"branch":"refs/heads/master"},"e11f86aecadd0a52eda4fbcdedd2ecfe6b5892b5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-05-21 05:41:03.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/70/792370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/70/792370/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/70/792370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/70/792370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/70/792370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/70/792370/3"}}},"commit":{"parents":[{"commit":"575086fd1fc12997036c4980d16c0f61a994a116","subject":"Use nova::compute::libvirt::virt_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/575086fd1fc12997036c4980d16c0f61a994a116"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-20 12:39:36.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2021-05-21 05:41:03.000000000","tz":0},"subject":"Add support for EL9 and python3.9","message":"Add support for EL9 and python3.9\n\nFollowing is done:-\n- Using distro module to get distribution as platform\n  module no longer provide that with python3.8+[1].\n- Use mariadb-server-galera as mariadb don\u0027t provide\n  mariadb-galera-server in EL9[2].\n- Ensure \u0027cronie\u0027 package is installed before attempting\n  start of crond service.\n\n[1] https://bugs.python.org/issue28167\n[2] https://src.fedoraproject.org/rpms/mariadb/c/b1bc71c1a\n\nChange-Id: I744de28a25739c00f585b7d9c12627ce0ed902f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e11f86aecadd0a52eda4fbcdedd2ecfe6b5892b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e11f86aecadd0a52eda4fbcdedd2ecfe6b5892b5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
