)]}'
{"id":"x%2Fpackstack~792787","triplet_id":"x%2Fpackstack~master~I4501db5c9d3d1e5155527d2a2dac7abd700ca6c8","project":"x/packstack","branch":"master","topic":"c9","hashtags":[],"change_id":"I4501db5c9d3d1e5155527d2a2dac7abd700ca6c8","subject":"Set ssl_verify to verify_peer with rabbitmq","status":"MERGED","created":"2021-05-24 08:16:26.000000000","updated":"2021-05-26 09:44:58.000000000","submitted":"2021-05-26 09:44:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792787-c9","meta_rev_id":"b8629d6f2668e5d7adf10bbbbbb14afce42b7b37","_number":792787,"virtual_id_number":792787,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-26 09:44:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2021-05-26 08:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-24 09:27:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2021-05-26 08:40:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-24 08:55:27.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2021-05-24 09:18:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-24 10:01:16.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-05-25 07:40:37.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"9bf4ba067272fcf3cb1550dc44e01ae63d583d34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-24 08:16:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae6009b494726c464d9a1e6827b3dc728a738d81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 09:18:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbf8bd5c8e714d49aed4299a57acbe80 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/948b7cd58d0c41e0a561857486120060 : SUCCESS in 4m 11s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/73d7cd8d69ef4bbaa1e9aceb556befc3 : SUCCESS in 53m 39s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/18af0f87539648ecb4e6c9a77626269d : SUCCESS in 40m 00s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1ff3acaf573f46a690bc16dace978a3b : SUCCESS in 50m 53s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/477e243053794993ab0b8b6b138dd01f : SUCCESS in 48m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d2571ed35395ce2285bd54616197391c7f652044","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-05-24 09:27:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"286ae721ee4f50762391db5dbde11b9a7c55bb64","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-05-26 08:40:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abbf657efb85b0ed22ecfe93837bda352865aa77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 08:41:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b5484c610fe210427b2694067619799ca011a8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 09:44:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c30d95e9fbd4b458afe3a4fce0ec792 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7696939664ed414da773f1ffeff65a9d : SUCCESS in 6m 44s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/01f1e54179144da7ab56d38f5c7bdf35 : SUCCESS in 59m 55s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4fbb476456c54cafb96c8e25ecf917db : SUCCESS in 43m 55s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0dba47a7946e4bed9c0b6ee258251160 : SUCCESS in 49m 55s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c2bd06712ca041709b21d95b2c71b45e : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b8629d6f2668e5d7adf10bbbbbb14afce42b7b37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 09:44:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f9f5b3adc67accbd66586d81442ded8e3b1ff23","revisions":{"0f9f5b3adc67accbd66586d81442ded8e3b1ff23":{"kind":"REWORK","_number":1,"created":"2021-05-24 08:16:26.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/87/792787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/87/792787/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/87/792787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/87/792787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/87/792787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/87/792787/1"}}},"commit":{"parents":[{"commit":"4dd2a7d0b23fba22e3db803b7116a3f930662781","subject":"Merge \"Use Stronger digest than sha1 for signing certs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4dd2a7d0b23fba22e3db803b7116a3f930662781"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-24 08:11:53.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-24 08:15:16.000000000","tz":330},"subject":"Set ssl_verify to verify_peer with rabbitmq","message":"Set ssl_verify to verify_peer with rabbitmq\n\nWith ssl_fail_if_no_peer_cert to true and\nssl_verify as \u0027verify_none\u0027 i.e default it\nno longer works with EL9, so let\u0027s set both.\n\nAlso as per doc[1] ssl_fail_if_no_peer_cert is used\nin conjuction with verify\u003dverify_peer.\n\n[1] https://erlang.org/doc/man/ssl.html\n\nChange-Id: I4501db5c9d3d1e5155527d2a2dac7abd700ca6c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0f9f5b3adc67accbd66586d81442ded8e3b1ff23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0f9f5b3adc67accbd66586d81442ded8e3b1ff23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
