)]}'
{"id":"x%2Fpackstack~795489","triplet_id":"x%2Fpackstack~master~I1bed7dc11425aee0ce7ae7e203eb53b88be3ddbe","project":"x/packstack","branch":"master","topic":"c9","hashtags":[],"change_id":"I1bed7dc11425aee0ce7ae7e203eb53b88be3ddbe","subject":"DNM test RDO pipeline running on c9","status":"ABANDONED","created":"2021-06-09 07:44:33.000000000","updated":"2022-12-16 13:30:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1f02850ed6ff187eff4b8ed55f909b9bee1b754e","_number":795489,"virtual_id_number":795489,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-09 08:49:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 08:49:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18637df1973f358ad7d83a14daf0c2d244315573","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-09 07:44:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa0e5d789bda2c5e056e339d1f002c072eebca82","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-09 08:09:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95aadfb1b699c61a00f575bf09620ea4f52ef130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 08:49:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f3b4610a43421a98384ef6ff503b5f : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a06cfcf27bf4861805e89eb718c2cb2 : SUCCESS in 5m 03s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/061128b7df524c59abc8d545de58a5ae : SUCCESS in 45m 37s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/39e9e37628e24d5cab6f6a9190dcb9ba : SUCCESS in 44m 07s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e7ab948dcba641d192ab00887ef31b8c : SUCCESS in 46m 07s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a189558ed31c4b5da109a29eccc492e6 : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f73022a55507ffec8c1f9608e42d238b4d42e963","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-09 12:15:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b23513174646778f006369b0c27d5337bae9d916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 13:10:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5add352b1ee41d9aca91bb56794f3ac : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8aa3a420e75a437c8f85df933ba8b7f8 : SUCCESS in 4m 30s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1b76afd18e694ae49a58757a3f7b040e : SUCCESS in 43m 06s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8acb4b8961c8466cb24f76cc476a5048 : SUCCESS in 45m 00s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/39093348380a4900b570924d36fe00ca : SUCCESS in 49m 53s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f1201298203644038acb662ff4dcf1aa : SUCCESS in 53m 57s","accounts_in_message":[],"_revision_number":1},{"id":"00565f347df1844f272342b479ecf3a3590840b2","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-09 15:20:36.000000000","message":"Patch Set 1:\n\ncheck rdo exeprimental","accounts_in_message":[],"_revision_number":1},{"id":"1f02850ed6ff187eff4b8ed55f909b9bee1b754e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-12-16 13:30:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f387f7ba93110812fa738c62554f5b2d9bd08a1","revisions":{"1f387f7ba93110812fa738c62554f5b2d9bd08a1":{"kind":"REWORK","_number":1,"created":"2021-06-09 07:44:33.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/89/795489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/89/795489/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/89/795489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/89/795489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/89/795489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/89/795489/1"}}},"commit":{"parents":[{"commit":"4f28c22b97b6c22a81868b0cfd8363df7fc76ad7","subject":"Remove deployment of panko","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f28c22b97b6c22a81868b0cfd8363df7fc76ad7"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-09 07:43:05.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-09 07:44:31.000000000","tz":120},"subject":"DNM test RDO pipeline running on c9","message":"DNM test RDO pipeline running on c9\n\nDepends-On: https://review.rdoproject.org/r/c/rdo-jobs/+/33995\nChange-Id: I1bed7dc11425aee0ce7ae7e203eb53b88be3ddbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1f387f7ba93110812fa738c62554f5b2d9bd08a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1f387f7ba93110812fa738c62554f5b2d9bd08a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
