)]}'
{"id":"x%2Fpackstack~800025","triplet_id":"x%2Fpackstack~master~Ib4c89772399de53d0dbdb8d11919e844e134191e","project":"x/packstack","branch":"master","topic":"dnmtrigger","hashtags":[],"change_id":"Ib4c89772399de53d0dbdb8d11919e844e134191e","subject":"[DNM] check auto trigger","status":"ABANDONED","created":"2021-07-08 12:29:16.000000000","updated":"2021-07-08 12:42:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"606f7d35c75d276757048a347fd21f4535a5a59b","_number":800025,"virtual_id_number":800025,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":0,"date":"2021-07-08 12:41:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2021-07-08 12:42:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2021-07-08 12:42:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-08 12:34:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d3fdeef2cca1e1af9f83841e556c32aea7ecc84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-08 12:29:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e0362840e02060738f43d4388a694584ffbc99e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 12:34:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c50a4eb34d4813949635b82c5c5b98 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2dfa9e0cb932ccf5b506e85be475a101820c404f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-08 12:40:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1980cde86cb535e2b4ffc40a481b445e41a7cec5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 12:41:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"254db49692271cad3ae80f6bbc7fdd164330231a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-08 12:42:10.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"606f7d35c75d276757048a347fd21f4535a5a59b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-08 12:42:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0749c339b580d7c5e36737e1a88c630179446a4b","revisions":{"0749c339b580d7c5e36737e1a88c630179446a4b":{"kind":"REWORK","_number":1,"created":"2021-07-08 12:29:16.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/25/800025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/25/800025/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/25/800025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/25/800025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/25/800025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/25/800025/1"}}},"commit":{"parents":[{"commit":"4f28c22b97b6c22a81868b0cfd8363df7fc76ad7","subject":"Remove deployment of panko","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f28c22b97b6c22a81868b0cfd8363df7fc76ad7"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-07-08 12:28:47.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-07-08 12:28:51.000000000","tz":330},"subject":"[DNM] check auto trigger","message":"[DNM] check auto trigger\n\nChange-Id: Ib4c89772399de53d0dbdb8d11919e844e134191e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0749c339b580d7c5e36737e1a88c630179446a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0749c339b580d7c5e36737e1a88c630179446a4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
