)]}'
{"id":"x%2Fpackstack~806105","triplet_id":"x%2Fpackstack~master~If05c4880dcd6ce4c61f43efbad67f59af7794981","project":"x/packstack","branch":"master","topic":"v2-api-removal","hashtags":[],"change_id":"If05c4880dcd6ce4c61f43efbad67f59af7794981","subject":"Drop usage of cinder v2 params","status":"MERGED","created":"2021-08-26 08:28:03.000000000","updated":"2021-08-26 15:54:55.000000000","submitted":"2021-08-26 15:54:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806105-v2-api-removal","meta_rev_id":"19aeadabbad12219c8b7d0a6c1c842ca8ab39a45","_number":806105,"virtual_id_number":806105,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-26 15:54:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2021-08-26 14:51:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-26 08:41:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2021-08-26 14:51:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 08:41:10.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2021-08-26 09:38:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 10:28:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"7174ffd21127e6f8c252b3b29b9880fc92388c35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-26 08:28:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99e32c78bac4a693389910d1607fd1375937be6b","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-08-26 08:41:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24c8640c5494e0c79a1e6993a25bfe8be3aec4ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 09:38:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48c02bd101e84cbc983817bacc86e7a5 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02fc82b1019346c1afa3c14089c04ab3 : SUCCESS in 4m 23s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/fd6927ff91504db48e16ea9de196334e : SUCCESS in 52m 13s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/659ebb8fb38143b7a500aee41fdc8c6a : SUCCESS in 52m 11s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/065c6b6e36504daa9f27d70c0ce2989c : SUCCESS in 50m 35s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4058e5a9e91b4abb97ba36f2e749b9a0 : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5e05118225a098ea86dd5d54da22b1809cdcb4d7","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-08-26 14:51:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"63ac791cd6d8ec710a1d89b925e2876db08df40a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 14:52:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16cbcc074cf60097ea4d2addc9cc9151d0811d8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 15:54:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a529636708e8449b8ad6fcc5f3ea3f28 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b659ee43f9f4cf8ba5b2a134b17bbf6 : SUCCESS in 6m 11s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4b8705ef864e499ebd5cdd55e0960e9c : SUCCESS in 53m 29s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7cdf1e75d64f43f3b2b52074d3ff72b8 : SUCCESS in 45m 26s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8f85e729ce9d4b0383ced369cfaa8b75 : SUCCESS in 55m 10s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ef8010d6ecf64d8d96aceeb1e98756fd : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":1},{"id":"19aeadabbad12219c8b7d0a6c1c842ca8ab39a45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 15:54:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3bc4d7e722710d2f2aeac11560ea997dcf899169","revisions":{"3bc4d7e722710d2f2aeac11560ea997dcf899169":{"kind":"REWORK","_number":1,"created":"2021-08-26 08:28:03.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/05/806105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/05/806105/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/05/806105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/05/806105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/05/806105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/05/806105/1"}}},"commit":{"parents":[{"commit":"d100111965a487e2feaa5210d5ff2be2fbec77fd","subject":"Include placement::api class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d100111965a487e2feaa5210d5ff2be2fbec77fd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-08-26 08:27:10.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-08-26 08:27:13.000000000","tz":330},"subject":"Drop usage of cinder v2 params","message":"Drop usage of cinder v2 params\n\nThese params were deprecated and are now removed\nwith [1], let\u0027s clear it\u0027s usage from packstack.\n\n[1] https://review.opendev.org/c/openstack/puppet-cinder/+/803874\n\nChange-Id: If05c4880dcd6ce4c61f43efbad67f59af7794981\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3bc4d7e722710d2f2aeac11560ea997dcf899169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3bc4d7e722710d2f2aeac11560ea997dcf899169"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
