)]}'
{"id":"x%2Fpackstack~813012","triplet_id":"x%2Fpackstack~master~I3bf1b93830e329ddcb9eb35695a0f145d459f9dd","project":"x/packstack","branch":"master","hashtags":[],"change_id":"I3bf1b93830e329ddcb9eb35695a0f145d459f9dd","subject":"Run packstack-upload-git-mirror on pre-release pipeline","status":"MERGED","created":"2021-10-07 13:58:06.000000000","updated":"2021-10-07 16:58:07.000000000","submitted":"2021-10-07 16:58:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813012","meta_rev_id":"41798607bc50543ae84f17ce9b907e1afdbb2e37","_number":813012,"virtual_id_number":813012,"owner":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-07 16:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2021-10-07 15:45:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2021-10-07 15:45:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 13:59:35.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-10-07 13:59:35.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-10-07 13:59:35.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2021-10-07 15:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc889d73bd2d5b33d3404395104c3589545db0bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-10-07 13:58:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61cfabe1f63dee66360fcb4aef3faee1a235251a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 15:11:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/264a142369424be79abe584bfe3609b9 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68175ee4578149179d0662f4b0d1f869 : SUCCESS in 5m 40s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0d6edcec79dd40369b1e3fbebace33ef : SUCCESS in 54m 10s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/46895bf24dba4d9aa8a0d69989b051d3 : SUCCESS in 56m 01s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/47b558e946774dbb9ecdbd9fee013a1c : SUCCESS in 45m 18s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1753298692594d50a6ba9078c8ead05c : SUCCESS in 1h 06m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4e5aebb103b6c570ea7a64367382e6af3d23f50a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-10-07 15:45:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cdd81e7792154263c098f6ffd3b7ee5751130326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 15:46:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3746f5fd505f518327fe84e1e9a4f140ebdb474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 16:58:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba57c189e6b24bccbc33bbabf364167c : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3245b16b3119479b9ee9942ce3855669 : SUCCESS in 5m 21s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0b0146f221f54d6a9e9fd53a58cc1d98 : SUCCESS in 53m 35s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4226c0db454c48caa0ef976665823cb9 : SUCCESS in 55m 20s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/35bfbce660944a69845127413d6a4cea : SUCCESS in 54m 47s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3f5f431c1d4d490bbbc1a5c74cebefb6 : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":1},{"id":"41798607bc50543ae84f17ce9b907e1afdbb2e37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 16:58:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"13d35f6f2f3aa5b34aac792841177b3cb080922b","revisions":{"13d35f6f2f3aa5b34aac792841177b3cb080922b":{"kind":"REWORK","_number":1,"created":"2021-10-07 13:58:06.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/12/813012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/12/813012/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/12/813012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/12/813012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/12/813012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/12/813012/1"}}},"commit":{"parents":[{"commit":"fbc954a97d7aa48c6da26fa4a3b51a84a6ec16ad","subject":"Merge \"Add CentOS Stream 9 preview support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fbc954a97d7aa48c6da26fa4a3b51a84a6ec16ad"}]}],"author":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-10-07 13:57:08.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2021-10-07 13:57:08.000000000","tz":120},"subject":"Run packstack-upload-git-mirror on pre-release pipeline","message":"Run packstack-upload-git-mirror on pre-release pipeline\n\nWe run it on the release pipeline, but not on the pre-release one, so\nnew RC tags do not get synchronized until a new commit is merged or a GA\ntag is pushed.\n\nChange-Id: I3bf1b93830e329ddcb9eb35695a0f145d459f9dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/13d35f6f2f3aa5b34aac792841177b3cb080922b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/13d35f6f2f3aa5b34aac792841177b3cb080922b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
