)]}'
{"id":"x%2Fpackstack~814585","triplet_id":"x%2Fpackstack~master~I151fba09ee25f28890aa7cb58fe2aa4013397c93","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-16 15:54:14.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I151fba09ee25f28890aa7cb58fe2aa4013397c93","subject":"Gnocchi: Remove unnecessary definition of [statsd] flush_delay","status":"MERGED","created":"2021-10-19 14:06:17.000000000","updated":"2025-03-16 17:04:14.000000000","submitted":"2025-03-16 17:04:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814585","meta_rev_id":"435003434d5252d445ca1cd57fdc4bf0a9ac97fe","_number":814585,"virtual_id_number":814585,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-16 17:04:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2025-02-19 22:09:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-03-12 08:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-16 14:19:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-03-12 08:43:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-19 15:07:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-19 15:11:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-19 22:09:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-03-12 08:43:05.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2025-03-16 14:19:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"2c81a4d91bdc2ec5eda01930a7a1716b86a02772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-19 14:06:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8b545022e9a8fb31e720cbfda2e0cb56afe102d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-10-19 15:07:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- packstack-centos9-integration-scenario001 https://review.rdoproject.org/zuul/build/11556fc0f4ca4419860fbc88cf0d552d : FAILURE in 26m 51s (non-voting)\n- packstack-centos9-integration-scenario002 https://review.rdoproject.org/zuul/build/7a0bcf703f22445d842569339e8e0079 : FAILURE in 47m 58s (non-voting)\n- packstack-centos9-integration-scenario003 https://review.rdoproject.org/zuul/build/1907cca6bb2a4c3a9dd4b235712da056 : FAILURE in 21m 33s (non-voting)\n- packstack-centos9-multinode-scenario002-tempest https://review.rdoproject.org/zuul/build/d3427c1b156f4ee6a8d986f0160876b2 : FAILURE in 59m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b796ac0ef30bd36dd5f03335a1eb37dad42e465b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 15:11:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5ad53f7adb247ac8a2035ec95775e78 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ba4c763e39b4e57b0a84ce668c2f792 : SUCCESS in 4m 19s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4ae1273c25c34bfb86813392ff175707 : SUCCESS in 42m 39s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7bf78f5244894b4fbf370c2209a65030 : SUCCESS in 57m 08s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3ebb8fe67c8d4552961acdcdb9c36bf4 : FAILURE in 18m 12s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/443de03b72f4490581b9c21db839d8b7 : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":1},{"id":"373782bd988dda0f4a5560232f948a826a591645","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-20 15:04:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2b11090e340d1e4a8d28dafdf8ed6003538f10b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-10-20 15:44:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- packstack-centos9-integration-scenario001 https://review.rdoproject.org/zuul/build/e18ee677b17b447a89364d87bc3978d1 : FAILURE in 25m 37s (non-voting)\n- packstack-centos9-integration-scenario002 https://review.rdoproject.org/zuul/build/2595aa456c2f4c3e92104ebc176ce287 : FAILURE in 36m 00s (non-voting)\n- packstack-centos9-integration-scenario003 https://review.rdoproject.org/zuul/build/7bea10add2574d2cbf6c081f7e7374b4 : FAILURE in 9m 57s (non-voting)\n- packstack-centos9-multinode-scenario002-tempest https://review.rdoproject.org/zuul/build/75b39278e9014f5ea1df5e0fcba8c79b : FAILURE in 38m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a52a20c0f2eea47a45700a1b14524d5a684b0161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 15:55:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b81795398854866932729dc74b7a99a : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a6099d6cb9d49189fd6d90574b67bba : SUCCESS in 5m 24s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/412131de7cd4460d8fea2d6a73fa3f97 : SUCCESS in 43m 36s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/112e39965d4a4ab49e93d235ecf51991 : SUCCESS in 46m 49s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ed149e43f8fc401e88dbca9eb835e868 : FAILURE in 13m 21s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4463fc3f343b48d685f14dff7a3fcef2 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e438118e19e047f6f9aef82ec0cca1548128adf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 14:48:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d6a0c3529960406a9d00b8bf06193e4baf7484ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 16:15:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5852c0486c2f41bc8b4f7be2b11d6912\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/494b7311ae654164885c00c3b155da24 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2f23d6b3279440ea96131346bc70fac : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6dc1fb9d70b4ad4a9e1c7baf9ab5ecc : SUCCESS in 3m 24s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/807032906508402a9d6a4f907b5462fa : SUCCESS in 58m 57s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/774d78908790440ca1dee97d3da4eb88 : SUCCESS in 1h 04m 10s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e30f925059e148af90f7af7ec0e23088 : SUCCESS in 1h 02m 23s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ccbd9166bd26414caed2b37d3d98387a : SUCCESS in 1h 21m 27s","accounts_in_message":[],"_revision_number":2},{"id":"74d866424b9389e1ffa689a7e7f21bc571fa3938","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-19 22:09:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"091c6c32a2708b6d9bc56f152dea8e150899a872","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2025-03-12 08:43:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e0d488c6682088893853eb12772615280f61e5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 08:43:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"61a966719c93ee7a38fb0933ba8a24bd078a433c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 09:31:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dbe591b983494aa283fd282c7975cc4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cdee25ce3ff4c008a69472cd651b740 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f90fe43a39949648a3cb9bdf077428b : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ffe54f77ca6a476da20cac8b7d42b802 : SUCCESS in 3m 19s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/cf96dec0292b445ea8d6709b0ca66b5d : SUCCESS in 40m 46s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5d2051aa03e5409bace60a0b132a0905 : SUCCESS in 24m 30s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3a84a1cb5220404d844ead79cc1d5152 : SUCCESS in 23m 42s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/187cabc4d7cb4f8b8cca63aed4185610 : FAILURE in 15m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f4a9b160e0c1ad686648ded0f3c4b45bd0e97a0d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-12 15:06:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6d2a53c22d47ed8e7ad710859893d4d196fbfe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 16:21:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c55977b0d514d048601a56cc8ffb81d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b550b869008f40e19d6b036ef1c6302d : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1211520e70c7402ab908c089789876f9 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9509219919434d5eb68240f5a3dbbd29 : SUCCESS in 4m 26s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/df75c7aec3f64ec2b7801697531a89b0 : SUCCESS in 1h 01m 44s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f631b2fed09843c38a6c0cbdf39423af : SUCCESS in 59m 08s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7ec6ec7fe6b54e97a53be75de01e349e : SUCCESS in 1h 08m 08s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5593e77ecf5d4940bf538dc161fe53a8 : FAILURE in 20m 58s","accounts_in_message":[],"_revision_number":2},{"id":"29a29f6d093b506da655a63705efbe1eeb63b6c0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-15 09:15:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"37a99f8a56dfc5ca4cd7ae16d57e7a0694d6276b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-15 10:01:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccadfbf6853e42a2a9fb1239ebb9b30d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a25ecc000c4912b92f9f91463ff639 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e9347b39cf54768959551f01b7bc52d : SUCCESS in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6457c8a1f9e543f9a0abb7b4ad0887c3 : SUCCESS in 6m 37s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/654b82f82c514e2b8e17985b1d3acd43 : SUCCESS in 39m 44s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/79be8aa15ab744858cc6c9b3cac0ecd5 : SUCCESS in 31m 08s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f8599c9018ad4774b7b61a020854ee7f : SUCCESS in 28m 47s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e62a59a13ab7484791f29ebf884c79e3 : FAILURE in 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"9e1f90fc6a83726c81fd90b2be93b36922515bb4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-16 06:21:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ebdd2dd6aea38f5fd973d453b719193bec3b2bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 06:54:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/216c86b188bd430ea417a69c54fbf73c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf89db22a9a845d4acd9476c97f5de4b : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd53cd5ec9ff4af8ad8ae1a678aec7f0 : SUCCESS in 1m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a543e10f0d024fb386d884f954c8f3c1 : SUCCESS in 2m 33s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/56d5d1e72cdd4f1ca54932abfd1c060f : SUCCESS in 22m 21s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cdc58563bb2c4a48806db0809a01f38f : SUCCESS in 27m 00s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/99303aa9e21a4afeba8833407d400ab2 : SUCCESS in 31m 55s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f502e7a9c500478fa72a55512f8975d1 : FAILURE in 16m 25s","accounts_in_message":[],"_revision_number":2},{"id":"e5c3675d569f570d3172bc1270bdeab122af1730","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-16 11:13:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1a5eaab00dcd778eb61b9364b657542dcefae5ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 12:15:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23bd5b78169742fab4cab252201a8aec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41cf7f26bd234ab8925254303be2f0fe : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58ffbbda2d124e8ea530b9106e74bf6a : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1217f5e36e18493bbe2240a3c2d9c4a9 : SUCCESS in 2m 54s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/64415a4e6b724ffe8a3a9cc3f886d395 : SUCCESS in 54m 17s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8284adfac3a14a2d904e9e3e088aa156 : SUCCESS in 28m 42s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/59cf34c442954b259b5fb71d2fdd9d5f : SUCCESS in 33m 29s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2d6914a40f934805a6ae7563dd828638 : SUCCESS in 31m 49s","accounts_in_message":[],"_revision_number":2},{"id":"d9eba0385787e5407b569ed6ff3adfb2909827c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 12:16:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bc7fb89e85b3404445ca4f02deebfebfca1bcf06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 13:06:54.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df4bbb94324543fa89b548fb0616913c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b48f2f7a29466d9038f16f7d4f4634 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ef9a977af9041b285576edf087088d5 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0ca180e908e47288cfd2f23447dcff9 : SUCCESS in 4m 28s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/02372c77d8224712a6b4988bae17f812 : SUCCESS in 43m 36s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cd38b5cbcad9447f92f075a1c3ad94e4 : SUCCESS in 25m 29s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/42482c9bab2747fc9482a31a7e16a02c : SUCCESS in 33m 43s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5c3243a60c0946a7823e55bc00d8275e : FAILURE in 21m 23s","accounts_in_message":[],"_revision_number":2},{"id":"3948e3933df1a623fa14a9b8e72e4c91d904e3d1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-16 14:19:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8d5afea5e164b5d364e004c231839966a366bb8d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-16 14:19:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48ad7e397299144299f004a45b7b73c9dd36879e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 14:55:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/587761983f4248df9ee1d571832fda57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53291aaa04ab4288af45ac6da554a62c : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74896f03133b4ec7adfc228c07c5839d : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/037d64b80a7a40a4a6bd66b6019ca28a : SUCCESS in 2m 50s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/65e0295bb16b419f922254f998099330 : SUCCESS in 34m 43s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5b3f1bd4ca8c466dbc8b6c16926f158a : SUCCESS in 31m 12s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2fdf1cc3dcc84aec8d3677d0398afdd8 : SUCCESS in 24m 12s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/425e182675cb495ea3e0ace9eb59647d : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f2605826b2ba83a1013265263b39bd6936150219","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 14:55:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d226d4c6344357ed428f296bd4033b9aec903b74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 15:53:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c67ae816b39458f986235606d45e4b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac1d1b2491bc4b0daaec98887c706dec : SUCCESS in 1m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23b06c0247364a05bb68fe0df1dc4cfe : SUCCESS in 1m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f5bb53714de4258992e66336ca63c86 : SUCCESS in 2m 18s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ac0462fdb6814aceb68a404273076e27 : SUCCESS in 23m 40s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2cd31a237e35475c8edd85a49893d897 : SUCCESS in 51m 16s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2f2a17a6db4845c3aaf17b0fb7c24630 : SUCCESS in 28m 21s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/dc4ab19da61340a3b9be36ddf514e10d : FAILURE in 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e21415dcaa896f95f82404965bc12636acd94dad","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-16 15:54:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f33195a27a4a333b3df7d051afc5a08916e8a5ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 16:28:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06dc8ab731854958bfe7a7ec47c1c967\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00811e06aea64eebb8df3dd360e8b953 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70cc5426dbc4454ab3b1a2feb98bd5fb : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28990ed73412412a8b60739aa6c67b0b : SUCCESS in 3m 07s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5464d2e76ecc49aaaa6670a20109a69c : SUCCESS in 28m 39s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d60878bd78d94ebc89353ada3e34fea2 : SUCCESS in 26m 26s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ae2a51c102b140998bf0a760ddd07b17 : SUCCESS in 32m 34s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0832a0fbeaec4585bd08952d2db4097f : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":2},{"id":"6650bec018e7c5014868832ab00d95d2ff094504","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 16:29:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e8ce8cd3144ca688cbba8eb3fe2dbbfa79d4c516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 17:04:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bca7a46558e649c5a5c09841aa9bd1cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f06c14d491164186988c2ba91bde4b6b : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0591bbd4e418486e8229e3b59426e39f : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6fea6c9e038e401190362edbff82e3cc : SUCCESS in 2m 43s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1136dfc5c3a9416d91994a97cc05f8b7 : SUCCESS in 23m 46s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4d004c7c366541479493c08e6ad96d5f : SUCCESS in 28m 47s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/346f0b95e4a24319a0bde04192d3ee35 : SUCCESS in 34m 51s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/71b1dea1b58140ca950a5f66d2fb5558 : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":2},{"id":"435003434d5252d445ca1cd57fdc4bf0a9ac97fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-16 17:04:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9813cdeed261cbc1150d123198c0f4b4d6cd004b","revisions":{"73633f80e2b70ca2379a3f3505d213698cbaf2ff":{"kind":"REWORK","_number":1,"created":"2021-10-19 14:06:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/814585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/85/814585/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/85/814585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/85/814585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/85/814585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/85/814585/1"}}},"commit":{"parents":[{"commit":"13d35f6f2f3aa5b34aac792841177b3cb080922b","subject":"Run packstack-upload-git-mirror on pre-release pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/13d35f6f2f3aa5b34aac792841177b3cb080922b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-19 14:05:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-19 14:05:07.000000000","tz":540},"subject":"Gnocchi: Remove unnecessary definition of [statsd] flush_delay","message":"Gnocchi: Remove unnecessary definition of [statsd] flush_delay\n\n... because the flush_delay parmaeter is no longer required.\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-gnocchi/+/814584\nChange-Id: I151fba09ee25f28890aa7cb58fe2aa4013397c93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/73633f80e2b70ca2379a3f3505d213698cbaf2ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/73633f80e2b70ca2379a3f3505d213698cbaf2ff"}]},"branch":"refs/heads/master"},"9813cdeed261cbc1150d123198c0f4b4d6cd004b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-18 14:48:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/814585/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/85/814585/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/85/814585/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/85/814585/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/85/814585/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/85/814585/2"}}},"commit":{"parents":[{"commit":"9073cffbd530b62aab18ee2350bd0bdd7fec565e","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9073cffbd530b62aab18ee2350bd0bdd7fec565e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-19 14:05:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 14:48:38.000000000","tz":0},"subject":"Gnocchi: Remove unnecessary definition of [statsd] flush_delay","message":"Gnocchi: Remove unnecessary definition of [statsd] flush_delay\n\n... because the flush_delay parmaeter is no longer required.\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-gnocchi/+/814584\nChange-Id: I151fba09ee25f28890aa7cb58fe2aa4013397c93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9813cdeed261cbc1150d123198c0f4b4d6cd004b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9813cdeed261cbc1150d123198c0f4b4d6cd004b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
