)]}'
{"id":"x%2Fpackstack~823373","triplet_id":"x%2Fpackstack~stable%2Fwallaby~I744de28a25739c00f585b7d9c12627ce0ed902f2","project":"x/packstack","branch":"stable/wallaby","topic":"cs9-wallaby","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-01-04 12:35:44.000000000","reason":"yatin replied on the change"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-01-04 13:42:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I744de28a25739c00f585b7d9c12627ce0ed902f2","subject":"Add support for EL9 and python3.9","status":"MERGED","created":"2022-01-04 10:32:38.000000000","updated":"2022-01-04 13:42:17.000000000","submitted":"2022-01-04 13:42:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823373-cs9-wallaby","meta_rev_id":"c0dfe63c7db0496eb8206e5108ffc338884ddcee","_number":823373,"virtual_id_number":823373,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-04 13:42:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-01-04 12:35:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-01-04 12:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2022-01-04 12:35:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-04 10:32:38.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-01-04 11:03:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 11:11:56.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2022-01-04 12:31:55.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"c7bc610d796ebb1e01b732c53b4fed7232255384","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-01-04 10:32:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"993c58e84e87c39cd0ab3411c66e3fdf505dc68a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 11:03:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c2935f4b7334eedb3369df8df09a043 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8aa16c50d5c480a8dd658597723150e : SUCCESS in 4m 57s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f933d08128124e339a0e44705dbcbf96 : POST_FAILURE in 21m 35s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/71aed5bd95784cf6ae60dcd45a0c4e63 : POST_FAILURE in 15m 15s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7ad20b949ff6470d9bc2a47b3577a898 : POST_FAILURE in 21m 32s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5cb1d99437934071934eb75c45083116 : POST_FAILURE in 16m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f827987f54ff5e1d483306b45298256e2527771a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-04 11:11:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e56aaa4ca3eb4fc20b5b086bff3fd9458b560b2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-01-04 11:20:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"09c0dfd018599eea5d2ba8d46ca87b32c6d17829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 12:26:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff2ce4e8d3cf4689a5ba19d781357c5c : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17a133d0117e4fd98ce70462c4ea07f9 : SUCCESS in 4m 11s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/40c9143b31fd4d26a784854d91824b1d : SUCCESS in 52m 55s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cbc61858668645d6b4ca08dac20c6fa5 : SUCCESS in 52m 24s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d448ae6b607c454ebcbb487118143c24 : SUCCESS in 56m 30s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c810acfa9042479db0f1fb46b8ab1cf4 : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":2},{"id":"3e38a39b20c9c3a5e9b27c116e3587fef95b4b31","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-04 12:31:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a3fafe54a95f370014798ac3e44bb6650b7c5426","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-01-04 12:35:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"db614ca664bd7c67c4c9b8f8654485919f723f5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 12:36:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8928772e751cac190da19645eed9ec6d3a53d7d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 13:42:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4829c34878ab4a92a87587b445c0fe7b : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90dfd359ce7c487abcd28f22ce43428c : SUCCESS in 4m 48s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2c03d981475d470f804b63b298aa990a : SUCCESS in 52m 41s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e35d27d5b8984fd4af2f3553ec42165f : SUCCESS in 52m 26s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/08a0470e20f64d2b9b18c1bf2252b866 : SUCCESS in 59m 43s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/fcae4513c34f4f3ebc7398abf9c56699 : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c0dfe63c7db0496eb8206e5108ffc338884ddcee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 13:42:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44df9d617e28578234bccbcbecc45be967046c4d","revisions":{"7408f47439d735d736e2722e736a92ab988d44a7":{"kind":"REWORK","_number":1,"created":"2022-01-04 10:32:38.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/73/823373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/73/823373/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/73/823373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/73/823373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/73/823373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/73/823373/1"}}},"commit":{"parents":[{"commit":"78b82ba6658fa19aa1556531ff91623d82e3d35d","subject":"Merge \"Remove deprecated keystone::ldap parameters\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/78b82ba6658fa19aa1556531ff91623d82e3d35d"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-20 12:39:36.000000000","tz":330},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-01-04 10:32:36.000000000","tz":60},"subject":"Add support for EL9 and python3.9","message":"Add support for EL9 and python3.9\n\nFollowing is done:-\n- Using distro module to get distribution as platform\n  module no longer provide that with python3.8+[1].\n- Use mariadb-server-galera as mariadb don\u0027t provide\n  mariadb-galera-server in EL9[2].\n- Ensure \u0027cronie\u0027 package is installed before attempting\n  start of crond service.\n\n[1] https://bugs.python.org/issue28167\n[2] https://src.fedoraproject.org/rpms/mariadb/c/b1bc71c1a\n\nChange-Id: I744de28a25739c00f585b7d9c12627ce0ed902f2\n(cherry picked from commit e11f86aecadd0a52eda4fbcdedd2ecfe6b5892b5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7408f47439d735d736e2722e736a92ab988d44a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7408f47439d735d736e2722e736a92ab988d44a7"}]},"branch":"refs/heads/stable/wallaby"},"44df9d617e28578234bccbcbecc45be967046c4d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-04 11:20:47.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/73/823373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/73/823373/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/73/823373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/73/823373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/73/823373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/73/823373/2"}}},"commit":{"parents":[{"commit":"f6ed301eba05f24e693b77a12a6af8eade6500c0","subject":"Move puppet-rsync to main branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f6ed301eba05f24e693b77a12a6af8eade6500c0"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-20 12:39:36.000000000","tz":330},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-01-04 11:20:13.000000000","tz":60},"subject":"Add support for EL9 and python3.9","message":"Add support for EL9 and python3.9\n\nFollowing is done:-\n- Using distro module to get distribution as platform\n  module no longer provide that with python3.8+[1].\n- Use mariadb-server-galera as mariadb don\u0027t provide\n  mariadb-galera-server in EL9[2].\n- Ensure \u0027cronie\u0027 package is installed before attempting\n  start of crond service.\n\n[1] https://bugs.python.org/issue28167\n[2] https://src.fedoraproject.org/rpms/mariadb/c/b1bc71c1a\n\nChange-Id: I744de28a25739c00f585b7d9c12627ce0ed902f2\n(cherry picked from commit e11f86aecadd0a52eda4fbcdedd2ecfe6b5892b5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/44df9d617e28578234bccbcbecc45be967046c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/44df9d617e28578234bccbcbecc45be967046c4d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
