)]}'
{"id":"x%2Fpackstack~823379","triplet_id":"x%2Fpackstack~stable%2Fwallaby~Ib67212dba7d550bf648f96a28ee9a4b3238eb405","project":"x/packstack","branch":"stable/wallaby","topic":"cs9-wallaby","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-01-04 12:30:31.000000000","reason":"Alfredo Moralejo replied on the change"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-01-04 13:42:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib67212dba7d550bf648f96a28ee9a4b3238eb405","subject":"Move puppet-rsync to main branch","status":"MERGED","created":"2022-01-04 11:19:40.000000000","updated":"2022-01-04 13:42:16.000000000","submitted":"2022-01-04 13:42:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823379-cs9-wallaby","meta_rev_id":"15028526093df860ebe1d866d34854e029a481bb","_number":823379,"virtual_id_number":823379,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-04 13:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-01-04 12:30:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-01-04 12:30:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-04 11:19:40.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2022-01-04 12:23:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"217c845b9357327c6d9b86600664d521e8f6ca04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-01-04 11:19:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2247a9bae4320927eeaa7b00abe79ed075562a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 12:23:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5317434a2fef494f9b91408face082cc : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b31658bab0e94a42b3631b9a83cf25a4 : SUCCESS in 3m 37s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a2f8b5e5d23e46c990b08b4269965cb1 : SUCCESS in 54m 50s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/76dbd275aa53492e9cf43c4f7b463f57 : SUCCESS in 50m 08s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9a90e8ae520b47e09ea47207b54af01a : SUCCESS in 55m 26s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/84027d09fd3d4ff6a9adac71745d1fa1 : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":1},{"id":"788db2e3c3bce4e371d4215040264db65c4f1590","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-04 12:30:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff55a85583b5dec9d21893b193d87a3c8f4d411b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 12:31:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"afd0ae55b51a689d28ec5ddc8d2264244f1cd981","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 13:42:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3f20c0439a42b0963acd9f334ff4e8 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dec97c7544d74337a1e64fe820c723e1 : SUCCESS in 4m 15s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/cf5ad376b028409ab5f5fa5048cf99a4 : SUCCESS in 51m 04s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/64b9720275a647209a3f1eb83538bd0b : SUCCESS in 1h 05m 54s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/158f01be51cc44d9bfaa843289b605d4 : SUCCESS in 56m 22s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e0f4bfcff46c4e8abd70b5ad642d4265 : SUCCESS in 1h 00m 51s","accounts_in_message":[],"_revision_number":1},{"id":"15028526093df860ebe1d866d34854e029a481bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 13:42:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6ed301eba05f24e693b77a12a6af8eade6500c0","revisions":{"f6ed301eba05f24e693b77a12a6af8eade6500c0":{"kind":"REWORK","_number":1,"created":"2022-01-04 11:19:40.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/79/823379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/79/823379/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/79/823379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/79/823379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/79/823379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/79/823379/1"}}},"commit":{"parents":[{"commit":"78b82ba6658fa19aa1556531ff91623d82e3d35d","subject":"Merge \"Remove deprecated keystone::ldap parameters\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/78b82ba6658fa19aa1556531ff91623d82e3d35d"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-09-01 12:01:34.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-01-04 11:18:13.000000000","tz":60},"subject":"Move puppet-rsync to main branch","message":"Move puppet-rsync to main branch\n\nmaster branch has been renamed to main upstream.\n\nChange-Id: Ib67212dba7d550bf648f96a28ee9a4b3238eb405\n(cherry picked from commit 091b810f1ad9de6e1fdf584fdea84e16d9ad858b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f6ed301eba05f24e693b77a12a6af8eade6500c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f6ed301eba05f24e693b77a12a6af8eade6500c0"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
