)]}'
{"id":"x%2Fpackstack~823484","triplet_id":"x%2Fpackstack~master~Ic82817f4ab9114e0970cf5256c57cb67c3bea147","project":"x/packstack","branch":"master","topic":"fix-ceilometer","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-01-05 13:06:40.000000000","reason":"yatin replied on the change"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-01-05 14:30:45.000000000","reason":"Alfredo Moralejo replied on the change"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-01-05 15:33:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic82817f4ab9114e0970cf5256c57cb67c3bea147","subject":"Move to ceilometer::agent::service_credentials","status":"MERGED","created":"2022-01-05 08:28:49.000000000","updated":"2022-01-05 15:33:00.000000000","submitted":"2022-01-05 15:33:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823484-fix-ceilometer","meta_rev_id":"1d31ed2d760025a55d603c065569d8fdd0ddd0e1","_number":823484,"virtual_id_number":823484,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-05 15:32:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-01-05 13:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-01-05 14:30:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 09:40:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-05 12:56:53.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2022-01-05 12:56:53.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"8573173f46af08cdbc47046aa54b258740d833a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-05 08:28:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"697775b94747a0e1e410c08973162dede052b497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 09:40:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7c606e8e7ba4fa4b6b30550964c0ea4 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f11ab107ca304a2d91818a134cc141f4 : SUCCESS in 4m 36s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2dbe27c7437340cdb5ab9dd6888737c9 : SUCCESS in 43m 06s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cce57d84894640ffb59033b2a77c6ea6 : SUCCESS in 50m 34s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f14bdbf06d0e45a3a4c82bad45a5f49a : SUCCESS in 51m 38s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/19f117d832ef4a2787e8809572c7caf2 : SUCCESS in 51m 05s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/29796805854b44a488e5c70cd2e32410 : SUCCESS in 33m 14s (non-voting)\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/51de28f249154e92978c7ae6ae794e69 : SUCCESS in 41m 58s (non-voting)\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c930d008ace34a1d9b32c59e4118567f : SUCCESS in 32m 34s (non-voting)\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6825dc57c2cf4c7082cc12590273d5e7 : SUCCESS in 57m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f461e41b99bad6d1d6959a93e5952a169e704229","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-01-05 13:06:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f6dff4a1190cb2e52ca4f230bab297be9394182c","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-05 14:30:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9551a5d93ffa5612fc24b3dc2015eeec0992deda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 14:34:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e0e4d09e6281d41aced6cf28c58d729770712f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 15:32:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d03095e40fd54aa2966e2156f2e5c0e9 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4640b4b944924dcea816ae0506423c2f : SUCCESS in 4m 21s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/3a49ad5cdfd642d8a15f3b7071c2afd6 : SUCCESS in 45m 26s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/84a7a55ce37d4970860835cdd8117da0 : SUCCESS in 45m 14s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8e453ab22d4a4d34bd3932101d964c27 : SUCCESS in 51m 07s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e29269e69e844d2c8603c810eec0dc3c : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1d31ed2d760025a55d603c065569d8fdd0ddd0e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 15:33:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8679c8e4691ee128189aec9fa1e2832bcb2ba184","revisions":{"8679c8e4691ee128189aec9fa1e2832bcb2ba184":{"kind":"REWORK","_number":1,"created":"2022-01-05 08:28:49.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/84/823484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/84/823484/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/84/823484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/84/823484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/84/823484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/84/823484/1"}}},"commit":{"parents":[{"commit":"a67138b2711f85288b118c378925ec938990835f","subject":"Use \"neutron::server::notification::nova\" class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a67138b2711f85288b118c378925ec938990835f"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-01-05 08:27:06.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-01-05 08:28:47.000000000","tz":60},"subject":"Move to ceilometer::agent::service_credentials","message":"Move to ceilometer::agent::service_credentials\n\nClass ceilometer::agent::auth has been removed in [1].\n\n[1] https://review.opendev.org/c/openstack/puppet-ceilometer/+/823006\n\nChange-Id: Ic82817f4ab9114e0970cf5256c57cb67c3bea147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8679c8e4691ee128189aec9fa1e2832bcb2ba184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8679c8e4691ee128189aec9fa1e2832bcb2ba184"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
