)]}'
{"id":"x%2Fpackstack~823700","triplet_id":"x%2Fpackstack~master~I1f10105b0196f6bf6bf0fe19470801bf7aff49bb","project":"x/packstack","branch":"master","topic":"glance-multi_store","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-01-12 15:09:15.000000000","reason":"yatin replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-12 16:18:05.000000000","reason":"Change was submitted"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-01-12 16:18:05.000000000","reason":"Change was submitted"},"13294":{"account":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"last_update":"2022-01-12 16:18:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1f10105b0196f6bf6bf0fe19470801bf7aff49bb","subject":"Glance: Use multistore resources to set up backend","status":"MERGED","created":"2022-01-06 14:52:45.000000000","updated":"2022-01-12 16:18:05.000000000","submitted":"2022-01-12 16:18:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823700-glance-multi_store","meta_rev_id":"6d68c2342aae6727397a33b69909d5271aa5c7d1","_number":823700,"virtual_id_number":823700,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-12 16:18:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-01-12 15:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-01-11 09:33:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2022-01-12 15:09:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 15:20:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-11 09:33:53.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2022-01-11 09:34:04.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2022-01-12 14:50:23.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"f86b076dc5a9404adca9b98573c12dccb55f3b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-06 14:52:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd1c5170c934c60e7e80bdb47feeb2e62047a6ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 15:20:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dbc2d9ad0754cdcb47ff2ff12f6bc6e : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0605ee2ddd5c4dcf96fc9f01b7720d18 : SUCCESS in 4m 04s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/604d2aca92cd43c699b3af168dc9a35b : FAILURE in 18m 56s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/603335424fd04410a5671d176a71a42d : FAILURE in 14m 06s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d83cf994cadb461fb42d464a4bd931c8 : FAILURE in 19m 36s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b16176ac88db4a828e0dd07037718990 : FAILURE in 17m 49s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/617d36a9d75845e6b6f01950f0443f21 : FAILURE in 15m 00s (non-voting)\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0d37fae59d794c8b83854463329d79d2 : FAILURE in 21m 34s (non-voting)\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/af0846da1b51476496480e1e6ebe5113 : FAILURE in 14m 04s (non-voting)\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/68fc5a6499d640b39a4fac75c62cedbd : FAILURE in 22m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1571d20b59b354178265d6538f1bec9eb9fa5587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2022-01-10 08:51:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9caeda8925f99860f75b6cd5ced1b24435ff0db1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 09:47:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0edcd9d6ac56486e9e5290e3c85a5c79 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/807bdd10111346adb41d639a42a79e2b : SUCCESS in 4m 15s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5049d7a714584ca4a46dcef24d33574a : SUCCESS in 38m 46s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7092e4b7af534df58da1c9e2d2705eae : SUCCESS in 41m 06s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/94318cc6d62644f59f05960c3d1bf0be : SUCCESS in 43m 14s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d183541590764a5e94fc646bf90da3e2 : SUCCESS in 48m 23s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e013a516870448b59a7ff09c31fdf2eb : SUCCESS in 40m 51s (non-voting)\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/494185c0e8b942eeab01c4de5f3447eb : SUCCESS in 40m 05s (non-voting)\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6a4e7a378ec547d08b2078c85595c0aa : SUCCESS in 38m 31s (non-voting)\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8a4fc9bb87714dc7a750ac0e5d7cbf3d : SUCCESS in 42m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b5bded8ef0ed58943772f539daabb8f7ea3229e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-11 09:33:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b082139fef0511801112990ce37863074ad9c8af","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-01-12 15:09:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"29b222c5cf098a6a5f69e761ded23d3b7ddcf68e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:10:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f9b65dd29bb148918149fccf5a3c5a3489928928","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:18:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cf5ead4313b4ba8a0a70f2c67568a44 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d48e0ee14fe4ec5919e7746f4deed94 : SUCCESS in 5m 07s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/708a6b0414d34ccb93ee3832c50547ff : SUCCESS in 51m 00s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/200ee821b68e46978d8a10d768d58c2b : SUCCESS in 59m 51s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/4afc2f50f3674cdc9648c327c3c67268 : SUCCESS in 50m 13s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8dccba0be62c4edc9bd5ca878b6bac91 : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6d68c2342aae6727397a33b69909d5271aa5c7d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:18:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"25af876371676fdcc1f8df94af4f61ecd7895daf","revisions":{"13bf49c0630beca26d2ea14ea99af5a39e4cf45c":{"kind":"REWORK","_number":1,"created":"2022-01-06 14:52:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/823700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/00/823700/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/00/823700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/00/823700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/00/823700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/00/823700/1"}}},"commit":{"parents":[{"commit":"8679c8e4691ee128189aec9fa1e2832bcb2ba184","subject":"Move to ceilometer::agent::service_credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8679c8e4691ee128189aec9fa1e2832bcb2ba184"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 14:51:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 14:52:33.000000000","tz":540},"subject":"Glance: Use multistore resources to set up backend","message":"Glance: Use multistore resources to set up backend\n\nChange-Id: I1f10105b0196f6bf6bf0fe19470801bf7aff49bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/13bf49c0630beca26d2ea14ea99af5a39e4cf45c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/13bf49c0630beca26d2ea14ea99af5a39e4cf45c"}]},"branch":"refs/heads/master"},"25af876371676fdcc1f8df94af4f61ecd7895daf":{"kind":"REWORK","_number":2,"created":"2022-01-10 08:51:21.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/00/823700/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/00/823700/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/00/823700/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/00/823700/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/00/823700/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/00/823700/2"}}},"commit":{"parents":[{"commit":"8679c8e4691ee128189aec9fa1e2832bcb2ba184","subject":"Move to ceilometer::agent::service_credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8679c8e4691ee128189aec9fa1e2832bcb2ba184"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 14:51:23.000000000","tz":540},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2022-01-10 08:51:20.000000000","tz":0},"subject":"Glance: Use multistore resources to set up backend","message":"Glance: Use multistore resources to set up backend\n\nChange-Id: I1f10105b0196f6bf6bf0fe19470801bf7aff49bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/25af876371676fdcc1f8df94af4f61ecd7895daf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/25af876371676fdcc1f8df94af4f61ecd7895daf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
