)]}'
{"id":"x%2Fpackstack~824025","triplet_id":"x%2Fpackstack~stable%2Fwallaby~I18fc7f85d1649a47000fa1a3289ab449e94be608","project":"x/packstack","branch":"stable/wallaby","topic":"c9-stable/wallaby","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-01-11 12:23:12.000000000","reason":"yatin replied on the change"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-01-11 15:13:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I18fc7f85d1649a47000fa1a3289ab449e94be608","subject":"Use Stronger digest than sha1 for signing certs","status":"MERGED","created":"2022-01-11 12:12:01.000000000","updated":"2022-01-11 15:13:47.000000000","submitted":"2022-01-11 15:13:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824025-c9-stable/wallaby","cherry_pick_of_change":792589,"cherry_pick_of_patch_set":1,"meta_rev_id":"9f90bd953dc0190f0369197048cdfc5f1a2de7ce","_number":824025,"virtual_id_number":824025,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-11 15:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-01-11 12:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2022-01-11 14:06:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-11 12:12:01.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-01-11 13:25:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d219d501c435f0ae2469518f8a5a12f5a81fd2c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-11 12:12:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c1d2be89bb1b46091ffe320db35eb870ad95655a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-01-11 12:23:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"afeb52c15e7bf9570c2458fffdb29a4cf87e0afc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 13:25:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16ef5a357eeb4146aba3c972df127d45 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/518896e10b4e42c8a435e7a85178203d : SUCCESS in 4m 04s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2cc0df0110714de18c6c17dd76d9e466 : SUCCESS in 44m 35s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2f8f956703a042edaa369177c520795c : SUCCESS in 1h 00m 40s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/af51345730344da38c2a7287ffa217d6 : SUCCESS in 51m 53s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7161de19f9574307b6185c97fb8efda4 : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8385caf73bb8ba303b42a2b6b6aef0770e149bae","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-01-11 14:06:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8eca60d00e79b5fa0a8b4bfe8304c3caa109203b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 14:07:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b582aae95958387eaf4d2a82430065447f39d4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 15:13:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5543f6a287694b509e05d165cad6ffaf : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cb3aa7e511841fbab11beb5e7b5d4cc : SUCCESS in 4m 52s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a7bd35b9c1ca4875b5ceb75e0ebd3296 : SUCCESS in 1h 00m 27s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4bf8bbb715204763875ad3c62acf59f8 : SUCCESS in 47m 10s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d04b68a13151431cb72c06e357f59f8c : SUCCESS in 55m 51s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8f47797d2b9d408aad42953a0a26a643 : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9f90bd953dc0190f0369197048cdfc5f1a2de7ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 15:13:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47f637f1d49ca9ae1baf1dce91994db8966ae562","revisions":{"47f637f1d49ca9ae1baf1dce91994db8966ae562":{"kind":"REWORK","_number":1,"created":"2022-01-11 12:12:01.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/25/824025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/25/824025/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/25/824025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/25/824025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/25/824025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/25/824025/1"}}},"commit":{"parents":[{"commit":"e14be19d3763a0d623879cf48b15a32f681d8d2e","subject":"Add CentOS Stream 9 preview support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e14be19d3763a0d623879cf48b15a32f681d8d2e"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-21 12:28:54.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-01-11 12:12:01.000000000","tz":0},"subject":"Use Stronger digest than sha1 for signing certs","message":"Use Stronger digest than sha1 for signing certs\n\nWith OpenSSL 3.0[1] sha1 can\u0027t be used with default\nsecurity level, let\u0027s use stronger digest that\u0027s\nsupported i.e sha256.\n\n[1] https://github.com/openssl/openssl/commit/aba03ae\n\nChange-Id: I18fc7f85d1649a47000fa1a3289ab449e94be608\n(cherry picked from commit 3b1d3594163c72630f5b4d0b4f2fff53d1f73c24)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/47f637f1d49ca9ae1baf1dce91994db8966ae562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/47f637f1d49ca9ae1baf1dce91994db8966ae562"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
