)]}'
{"id":"x%2Fpackstack~824173","triplet_id":"x%2Fpackstack~master~I6425014a5e7f72a06da08405cddb44f906a6b7ac","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-01-12 14:09:12.000000000","reason":"yatin replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-12 16:34:08.000000000","reason":"Change was submitted"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-01-12 16:34:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6425014a5e7f72a06da08405cddb44f906a6b7ac","subject":"Magnum: Use specific hieradata for keystone urls","status":"MERGED","created":"2022-01-11 12:47:58.000000000","updated":"2022-01-12 16:34:08.000000000","submitted":"2022-01-12 16:34:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824173","meta_rev_id":"3346aecb4957c28570f85b0217b6d6a7b71315e0","_number":824173,"virtual_id_number":824173,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-12 16:34:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-01-12 14:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-01-12 15:22:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-01-12 15:22:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-11 14:02:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-12 07:29:05.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2022-01-12 07:29:16.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-01-12 07:29:16.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"71669a71ea5b28220eb9383930966f5fda8246b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-11 12:47:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d42cf2a1c5ad551d9a6263f73957840d677fa43b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 14:02:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a971b07881b2439abcc0cf2b0b04efb1 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a95fd9d09354fc5b24a207fe2ebf63a : SUCCESS in 4m 04s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1379ba9636ce4256bc3e1ae35376eb4c : SUCCESS in 51m 54s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/870f203e10f64f33b3e24d04603460d6 : SUCCESS in 51m 12s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/74a496fe95194b58abce83ac5d4caf75 : SUCCESS in 53m 11s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cc3d8ac44a5a46d094d6b60d277f4e6f : SUCCESS in 1h 00m 19s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d18111e9bb6b498ea96c4d0ee668368c : SUCCESS in 44m 37s (non-voting)\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/faab3466ba33472b83936d934403c7bd : SUCCESS in 49m 23s (non-voting)\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e9c73fdcf8904a40825580f66f2b83f7 : SUCCESS in 45m 24s (non-voting)\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1e4d3b31d0764df69476906f1b4c3ffa : SUCCESS in 58m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"739c5a009c61c2ff869d4b8e605af20dfdb4a94b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-12 07:29:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c53e87691cea29ba22bf58c2279155ddf827652","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 13:30:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a7c5e0a6d8fc5ab5744e95e82f9ea948b7d3ca4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 13:34:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c98513e4c7a18a99fbc8fc36c53c7615148c9796","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 13:50:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2c1495786448d951ba37705cdda6e9bcc7422e3","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-01-12 14:09:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a0eb69a6fe74f24e514908b8ca53046cef604740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:20:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96a781aa793b457e929b48e6f37187ea : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/395acda848ad4de6bcbb22e8a700afad : SUCCESS in 6m 10s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5ba3c8e0d8cc45a38e9d9f96ff90e85c : SUCCESS in 43m 24s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d4379069761443eb9dcd38a8d841dc4e : SUCCESS in 44m 25s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/50f0ee3593804e468b4949db2478a8c6 : SUCCESS in 1h 04m 43s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/daea85ae20414231ab9fe960b32cedeb : SUCCESS in 1h 01m 16s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d12e068940a9468cb2a6b8b26f04ac81 : SUCCESS in 47m 39s (non-voting)\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/124d30e2bad340e98dec89c2b36a5b59 : SUCCESS in 48m 31s (non-voting)\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9a5f09342ae14ad8a437e10967dfe69f : SUCCESS in 46m 52s (non-voting)\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e23d47c5b5544a43b603fa21a4f96d9d : SUCCESS in 56m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0427d780b224a81384399f03b11ce6975fa44b84","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-12 15:22:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c72bb28ff85e6e7b5d54740b11ce29d8e0c7a6af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:24:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"28a6782f7adc05ef53c4e9bc51232966f9d5f4f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:34:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d26d8b2fa2b43ff87e288a5e5e54e11 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/320b75692454452db7d9dd96fe1da6e4 : SUCCESS in 5m 17s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/fcc90324fef54e81921c91c9ed6bc294 : SUCCESS in 46m 24s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1f0bf48384024234afe5f32ab19ed9a7 : SUCCESS in 50m 53s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e8462f79bcc24e1b83fe2ed2d15e8036 : SUCCESS in 54m 49s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/18c5cec44f4e4fcd8933f7f642b08761 : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":2},{"id":"3346aecb4957c28570f85b0217b6d6a7b71315e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:34:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e88cad7a9f20d65c36d15f51f7821a355223292","revisions":{"ee6439e165d6bd488ae48d8568cf88feebc2d90b":{"kind":"REWORK","_number":1,"created":"2022-01-11 12:47:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/824173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/73/824173/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/73/824173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/73/824173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/73/824173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/73/824173/1"}}},"commit":{"parents":[{"commit":"8679c8e4691ee128189aec9fa1e2832bcb2ba184","subject":"Move to ceilometer::agent::service_credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8679c8e4691ee128189aec9fa1e2832bcb2ba184"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-11 12:46:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-11 12:46:51.000000000","tz":540},"subject":"Magnum: Fix urls used for authtoken middleware","message":"Magnum: Fix urls used for authtoken middleware\n\nThe keystone url parameters should be honored instead of the host where\nthe magnum services are running.\n\nChange-Id: I6425014a5e7f72a06da08405cddb44f906a6b7ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ee6439e165d6bd488ae48d8568cf88feebc2d90b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ee6439e165d6bd488ae48d8568cf88feebc2d90b"}]},"branch":"refs/heads/master"},"7e88cad7a9f20d65c36d15f51f7821a355223292":{"kind":"REWORK","_number":2,"created":"2022-01-12 13:34:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/824173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/73/824173/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/73/824173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/73/824173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/73/824173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/73/824173/2"}}},"commit":{"parents":[{"commit":"8679c8e4691ee128189aec9fa1e2832bcb2ba184","subject":"Move to ceilometer::agent::service_credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8679c8e4691ee128189aec9fa1e2832bcb2ba184"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-11 12:46:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 13:33:57.000000000","tz":540},"subject":"Magnum: Use specific hieradata for keystone urls","message":"Magnum: Use specific hieradata for keystone urls\n\n... instead of composing these urls from host.\n\nChange-Id: I6425014a5e7f72a06da08405cddb44f906a6b7ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7e88cad7a9f20d65c36d15f51f7821a355223292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7e88cad7a9f20d65c36d15f51f7821a355223292"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
