)]}'
{"id":"x%2Fpackstack~826476","triplet_id":"x%2Fpackstack~master~Ia59c0f11b9ce2eaf29cd23e8ca8de9af0d90f69f","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-01-27 10:01:02.000000000","reason":"Change was submitted"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-01-27 10:01:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia59c0f11b9ce2eaf29cd23e8ca8de9af0d90f69f","subject":"Move nova and aodh database connection config to ::db","status":"MERGED","created":"2022-01-26 14:38:38.000000000","updated":"2022-01-27 10:01:02.000000000","submitted":"2022-01-27 10:01:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"826476","meta_rev_id":"c1782c1dd5202d65f286037ed5feae6f83fd1a94","_number":826476,"virtual_id_number":826476,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 10:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-01-27 08:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-27 08:55:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-27 08:55:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-26 14:39:25.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2022-01-26 15:12:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-26 15:32:45.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"}],"messages":[{"id":"dd6f595831fe10296c0eec6f212079b79f6f5d63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-26 14:38:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7708c1646bc2175dcf75c8cb95013fe873639c3","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-01-26 14:39:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dc4e29beea64473898d72dbab999cb8dbceefc3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 15:12:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd2e2bd637f467cabd0e83b6026f92d : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a13bd2b924c4ad58fda793207da9648 : SUCCESS in 5m 57s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4f01425201254be58d000c072b335787 : FAILURE in 21m 29s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ab32c6b1cd264956a25fd54249de9653 : FAILURE in 14m 00s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/59dc52a833bb4d458bfce4822474699d : FAILURE in 19m 13s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5114d308e47042cb89ba7be8199880c3 : FAILURE in 21m 55s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/732a0398b0a24e9fac97182255628e57 : FAILURE in 22m 27s (non-voting)\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b2c6ddb4f66a43a28363732d9f525fb6 : FAILURE in 19m 39s (non-voting)\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c8bd73e0e8ec47928f95e9303e1ec1c4 : FAILURE in 18m 58s (non-voting)\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3a35c9c2f2da449696e1f19c3a6306c2 : FAILURE in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5ae916507cfc04766d313b050a2dff702bdfd8f","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2022-01-26 15:32:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"adc01aff6f0c3d6da2bed6e578bcfa81442632af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-26 15:40:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4df5a1ccb36d9392f65f3c1178668684b84108b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 16:46:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/240347915b7549da8d08b42a5941078c : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be29fd56304f4225bbb17f6a0eeb287e : SUCCESS in 4m 55s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/df74bde603944972b59f651e5db11148 : SUCCESS in 48m 37s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/860ea93e31a04e77879b156205cc4695 : SUCCESS in 54m 24s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/06b251cb75a448e7b4d2a2e3e1662d30 : FAILURE in 14m 06s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/034ad88aacfa4d55a09d249b3a3e5e84 : SUCCESS in 59m 28s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c6bff4c78f9949358708351ea20c05ea : SUCCESS in 32m 52s (non-voting)\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e1351d29beca47d6835ea7f2fc367e0d : SUCCESS in 45m 25s (non-voting)\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a20a0007efde4b7a8f0b4190b9990f7e : FAILURE in 13m 52s (non-voting)\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/156546d070c94caf9db7a12bba9cf2ac : SUCCESS in 43m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d22705da58c567c599818bbf7424325b36fb9b4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-01-26 17:08:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4fb029b0e72b563d617299244cc1b7029dc7c75a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 18:27:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d6739530dd0455ebb35015fa16bac0e : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac1543ffee874fe3a17aa3428cbe2dc7 : SUCCESS in 4m 09s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/31a6bbd46ad94a6ba2308ffc4029c3a3 : SUCCESS in 1h 12m 21s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6963e4972ed74c83ae56899bcb8b56da : SUCCESS in 42m 04s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9ef744082cee49f3a92cdddec0831b52 : SUCCESS in 1h 03m 16s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4e5c3fc4a8024025969481d08dae64c7 : SUCCESS in 48m 38s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/97b853ab530b40229a97cfe9edcb0641 : SUCCESS in 55m 49s (non-voting)\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/33cfd58d9eb04aa395ad939d04ec4c7e : SUCCESS in 48m 26s (non-voting)\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/95de8903c5ac4187a402eee9c4affe69 : SUCCESS in 44m 05s (non-voting)\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cde5a3c2bba24cc78edf434273ce6416 : SUCCESS in 43m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6de7bea493f9fb47cd0cee62f01f64fb86d57de1","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-01-27 08:39:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9b0c0e2b922c48c6b74f9508db58b4ea6ce55594","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2022-01-27 08:55:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8d79659aa40f90a2b6dae2973787b7cd35eeec79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 08:56:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"df296249a0c52d2ed92620d461ad7b14f03bd800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 10:01:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75898231489949f1817ed76ded1fef12 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cd00023b57649aab2e52b205d57fe6d : SUCCESS in 5m 21s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1e90d60776fb40dfb0408a81a4870ceb : SUCCESS in 43m 21s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/10634a40a268425e9fed63d0558de7d7 : SUCCESS in 48m 17s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/58c639f0cd0042f687c9c36f48c08398 : SUCCESS in 50m 43s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d019623b531e4e058399bf84d005bf88 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":3},{"id":"c1782c1dd5202d65f286037ed5feae6f83fd1a94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 10:01:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cfb5f0f9623a5eab44771912968836b4a489924e","revisions":{"8b212306071e41b80f67e2d2f40b058afddca815":{"kind":"REWORK","_number":1,"created":"2022-01-26 14:38:38.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/76/826476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/76/826476/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/76/826476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/76/826476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/76/826476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/76/826476/1"}}},"commit":{"parents":[{"commit":"69b7e1219a5bece8f1ba48c6b222e450306d2bd5","subject":"Merge \"Magnum: Use specific hieradata for keystone urls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/69b7e1219a5bece8f1ba48c6b222e450306d2bd5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-01-26 10:30:13.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-01-26 14:38:37.000000000","tz":60},"subject":"Move nova database connection config to nova::db","message":"Move nova database connection config to nova::db\n\nThe parameters have been removed from main nova class in [1].\n\n[1] https://review.opendev.org/c/openstack/puppet-nova/+/825960\n\nChange-Id: Ia59c0f11b9ce2eaf29cd23e8ca8de9af0d90f69f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8b212306071e41b80f67e2d2f40b058afddca815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8b212306071e41b80f67e2d2f40b058afddca815"}]},"branch":"refs/heads/master"},"fb17652f9f2d8587ba6c5304c0c62e45d355cfda":{"kind":"REWORK","_number":2,"created":"2022-01-26 15:40:15.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/76/826476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/76/826476/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/76/826476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/76/826476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/76/826476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/76/826476/2"}}},"commit":{"parents":[{"commit":"69b7e1219a5bece8f1ba48c6b222e450306d2bd5","subject":"Merge \"Magnum: Use specific hieradata for keystone urls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/69b7e1219a5bece8f1ba48c6b222e450306d2bd5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-01-26 10:30:13.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-01-26 15:40:08.000000000","tz":60},"subject":"Move nova database connection config to nova::db","message":"Move nova database connection config to nova::db\n\nThe parameters have been removed from main nova class in [1].\n\n[1] https://review.opendev.org/c/openstack/puppet-nova/+/825960\n\nChange-Id: Ia59c0f11b9ce2eaf29cd23e8ca8de9af0d90f69f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fb17652f9f2d8587ba6c5304c0c62e45d355cfda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fb17652f9f2d8587ba6c5304c0c62e45d355cfda"}]},"branch":"refs/heads/master"},"cfb5f0f9623a5eab44771912968836b4a489924e":{"kind":"REWORK","_number":3,"created":"2022-01-26 17:08:04.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/76/826476/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/76/826476/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/76/826476/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/76/826476/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/76/826476/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/76/826476/3"}}},"commit":{"parents":[{"commit":"69b7e1219a5bece8f1ba48c6b222e450306d2bd5","subject":"Merge \"Magnum: Use specific hieradata for keystone urls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/69b7e1219a5bece8f1ba48c6b222e450306d2bd5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-01-26 10:30:13.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-01-26 17:07:37.000000000","tz":60},"subject":"Move nova and aodh database connection config to ::db","message":"Move nova and aodh database connection config to ::db\n\nThose parameters have been removed from main nova class in [1],\nand aodh class in [2].\n\n[1] https://review.opendev.org/c/openstack/puppet-nova/+/825960\n[2] https://review.opendev.org/c/openstack/puppet-aodh/+/825961\n\nChange-Id: Ia59c0f11b9ce2eaf29cd23e8ca8de9af0d90f69f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cfb5f0f9623a5eab44771912968836b4a489924e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cfb5f0f9623a5eab44771912968836b4a489924e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
