)]}'
{"id":"x%2Fpackstack~827530","triplet_id":"x%2Fpackstack~master~Ie3eb1305fe375f706ba5dd7e8c84c5a53efbe714","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-02-03 05:07:48.000000000","reason":"yatin replied on the change"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-02-03 06:11:31.000000000","reason":"Change was submitted"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-02-03 06:11:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie3eb1305fe375f706ba5dd7e8c84c5a53efbe714","subject":"Move database connection config to ::db classes for glance and cinder","status":"MERGED","created":"2022-02-02 16:57:06.000000000","updated":"2022-02-03 06:11:31.000000000","submitted":"2022-02-03 06:11:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827530","meta_rev_id":"dd9466c1d4541007c8ef0428e5c1d5d168bc2c29","_number":827530,"virtual_id_number":827530,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 06:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-02-03 05:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2022-02-03 05:07:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-02 18:06:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-02 18:07:51.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2022-02-02 18:07:51.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"cd6ca00e531e04bd18453cf4bb96c2aebf1d0d4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-02 16:57:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d1c9d72aef5c54765f2ba7f37991355c84ef7c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 18:06:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dfc61e8466e4f019452b1405f0ca204 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54a1235731d3456192a65c59ea4533c7 : SUCCESS in 4m 38s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c2f10053602c4dfd951dc42f139b4c74 : SUCCESS in 47m 58s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d2bf9e221a4349d4944cdfbfa35b294e : SUCCESS in 46m 44s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/12c54214fabb4e5bb89d01c58e40850b : SUCCESS in 50m 23s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/05b7d2b807004fa092ef4987b3b78904 : SUCCESS in 54m 26s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4b033b2e654947dd9b0357038791fb14 : SUCCESS in 39m 46s (non-voting)\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/27a6e6476de24c4dbd6accf2a24891b4 : SUCCESS in 57m 39s (non-voting)\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e41a6e828a4f4e44ab7298e7253fb46f : SUCCESS in 48m 46s (non-voting)\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ced4e979393b4d9898d5868782f15acb : SUCCESS in 50m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef01dff24e371d846dc3ce30cd9cd203ed29d5f5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 05:07:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2474f13a09ef48104fbe04840eabf54d04974647","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 05:08:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b48361fb6ccf28cb4177a28a1148f719770d8684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 06:11:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4842680c162b40149c4a3c0e8d0abb60 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a55a30cd0d55415abc7d72bd5256caa8 : SUCCESS in 3m 45s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c4a7c0912aa54a4fba627288eb6684c2 : SUCCESS in 49m 04s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d496819a35f541ac9a225a2c620258e6 : SUCCESS in 58m 20s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a318e51aa27e4f308685c59a9b153293 : SUCCESS in 46m 34s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a1c27b4011414206997a4a5a5abd42b7 : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":1},{"id":"dd9466c1d4541007c8ef0428e5c1d5d168bc2c29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 06:11:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f69faf24acc947c0bfe4b1892cdb2bb628eec1a","revisions":{"2f69faf24acc947c0bfe4b1892cdb2bb628eec1a":{"kind":"REWORK","_number":1,"created":"2022-02-02 16:57:06.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/30/827530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/30/827530/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/30/827530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/30/827530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/30/827530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/30/827530/1"}}},"commit":{"parents":[{"commit":"cfb5f0f9623a5eab44771912968836b4a489924e","subject":"Move nova and aodh database connection config to ::db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cfb5f0f9623a5eab44771912968836b4a489924e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-02 16:50:24.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-02 16:56:49.000000000","tz":60},"subject":"Move database connection config to ::db classes for glance and cinder","message":"Move database connection config to ::db classes for glance and cinder\n\nThose parameters have been removed from main glance class in [1] and\nfrom cinder in [2].\n\n[1] https://review.opendev.org/c/openstack/puppet-glance/+/827031\n[2] https://review.opendev.org/c/openstack/puppet-cinder/+/827003\n\nChange-Id: Ie3eb1305fe375f706ba5dd7e8c84c5a53efbe714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2f69faf24acc947c0bfe4b1892cdb2bb628eec1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2f69faf24acc947c0bfe4b1892cdb2bb628eec1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
