)]}'
{"id":"x%2Fpackstack~827610","triplet_id":"x%2Fpackstack~master~I5cded1a355859dce02967ad1caf7df5e17058876","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-02-03 10:53:55.000000000","reason":"yatin replied on the change"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-02-04 06:11:57.000000000","reason":"Change was submitted"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-02-03 15:42:44.000000000","reason":"Joel Capitao replied on the change"},"13294":{"account":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"last_update":"2022-02-03 17:29:51.000000000","reason":"Javier Peña replied on the change"}},"hashtags":[],"change_id":"I5cded1a355859dce02967ad1caf7df5e17058876","subject":"Use puppet-passed-ci repos for CS9","status":"MERGED","created":"2022-02-03 09:26:58.000000000","updated":"2022-02-04 06:11:57.000000000","submitted":"2022-02-04 06:11:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827610","meta_rev_id":"9a42aad54858096954b52f6c289e39b409593ee2","_number":827610,"virtual_id_number":827610,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 06:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-02-04 05:15:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":2,"date":"2022-02-03 17:29:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2022-02-04 05:15:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 10:41:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 10:52:32.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2022-02-03 10:52:32.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2022-02-03 10:52:32.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"4686737cee46c214f145fcd46f18316efc6b58f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-03 09:26:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70e511bfa3713ffeba29de1568c94293467a7b04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 10:41:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1690e1e4634b467d9b6c0ee4429d6f57 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bda7026ec97f40048c9a6f91717e2a7b : SUCCESS in 4m 19s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2426fcec28504f4fa0d3a057c04308c2 : SUCCESS in 37m 40s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f9a0c1976f544c72851c5ab16fd9482f : SUCCESS in 49m 04s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b9684fcc0b3c49aa8030924c5cde1a27 : SUCCESS in 50m 56s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/de850b30d0d542b2b41da8d38e5275e1 : SUCCESS in 49m 21s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a103e20ae950491d9965847b12000d41 : SUCCESS in 44m 17s (non-voting)\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8d690b30c2814e49b9a1759a785e0566 : SUCCESS in 48m 25s (non-voting)\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/787a0ca5b72b47d09035325c9cbbea72 : SUCCESS in 47m 43s (non-voting)\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/576a639d051e4ae08eed51e399f96f8a : SUCCESS in 58m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2960404d618124cf89e3d724d5d04cd7ec8cbcce","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 10:53:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"86b993be97a136f62f65abaf63f2654e0603b05d","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-02-03 15:42:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6a098493c89042b35e9fc614507e7b96c7bc24f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-03 15:46:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ad695436b2fd22345ed4bb463e11ad7e5d74e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 17:04:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9974f00e593c416498e934a355e09839 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/252b6f31b0f2425cb71c189d48bd3503 : SUCCESS in 4m 13s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/346f1acd22054df4a9556a1176c11180 : SUCCESS in 52m 39s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d8eead12ceba423b8dd2ea7188f97cf6 : SUCCESS in 55m 59s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ff408be24caf43059dbe73b721aac890 : SUCCESS in 55m 49s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c69f1f3eaa434bcd9d9c0d5e1531d87d : SUCCESS in 58m 31s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/bb9fc7ee1e4149ee93092af2d5c7acc8 : SUCCESS in 56m 41s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5679e2647cb34f3b901e00d1ece5037a : SUCCESS in 44m 10s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a3f4ad5a888b427cb02ad09379937cd1 : SUCCESS in 1h 02m 38s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a8847e1ea5174a858a34cdbbb6ffc85b : SUCCESS in 56m 23s","accounts_in_message":[],"_revision_number":2},{"id":"4920281f9d24a9570f8dcd74ce3c9e577bf790e0","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2022-02-03 17:29:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e55d9aa821861a65fefdcaa00af66161640716fe","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-04 05:15:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c4094fa0dfb9aa544b9fc81d94fb217f7fc21e28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 05:16:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63cd9f2768a003594a7ae5b5a34311101e610ab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 06:11:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/336eb57c3ac94dd9a48408fa5c8dbb1e : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33df0e1a80414869aa8486a867e9aef5 : SUCCESS in 4m 04s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f3b15bf4c81a45b6bb71ed9a0b09f0ae : SUCCESS in 40m 46s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f405a306215749619946901ce1aaba51 : SUCCESS in 43m 21s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1d9a6ffa0eea4032b247651ce24222f4 : SUCCESS in 48m 06s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/54a7a56aab3f48f8ac8efac207838cab : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9a42aad54858096954b52f6c289e39b409593ee2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 06:11:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"500e6e6de934e4b9dc80d8c3feb0585aaa9c6a2e","revisions":{"9a9bd4136960c492da81f7e3902b83801e99a7b7":{"kind":"REWORK","_number":1,"created":"2022-02-03 09:26:58.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/10/827610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/10/827610/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/10/827610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/10/827610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/10/827610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/10/827610/1"}}},"commit":{"parents":[{"commit":"2f69faf24acc947c0bfe4b1892cdb2bb628eec1a","subject":"Move database connection config to ::db classes for glance and cinder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2f69faf24acc947c0bfe4b1892cdb2bb628eec1a"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-03 09:26:13.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-03 09:26:13.000000000","tz":60},"subject":"Use puppet-passed-ci repos for CS9","message":"Use puppet-passed-ci repos for CS9\n\nNow we have a puppet promotion pipeline in place, so let\u0027s use the last\npromoted repos instead of current ones.\n\nChange-Id: I5cded1a355859dce02967ad1caf7df5e17058876\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9a9bd4136960c492da81f7e3902b83801e99a7b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9a9bd4136960c492da81f7e3902b83801e99a7b7"}]},"branch":"refs/heads/master"},"500e6e6de934e4b9dc80d8c3feb0585aaa9c6a2e":{"kind":"REWORK","_number":2,"created":"2022-02-03 15:46:18.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/10/827610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/10/827610/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/10/827610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/10/827610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/10/827610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/10/827610/2"}}},"commit":{"parents":[{"commit":"2f69faf24acc947c0bfe4b1892cdb2bb628eec1a","subject":"Move database connection config to ::db classes for glance and cinder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2f69faf24acc947c0bfe4b1892cdb2bb628eec1a"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-03 09:26:13.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-03 15:45:54.000000000","tz":60},"subject":"Use puppet-passed-ci repos for CS9","message":"Use puppet-passed-ci repos for CS9\n\nNow we have a puppet promotion pipeline in place, so let\u0027s use the last\npromoted repos instead of current ones.\n\nWe are also making cs9 jobs voting.\n\nChange-Id: I5cded1a355859dce02967ad1caf7df5e17058876\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/500e6e6de934e4b9dc80d8c3feb0585aaa9c6a2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/500e6e6de934e4b9dc80d8c3feb0585aaa9c6a2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
