)]}'
{"id":"x%2Fpackstack~827697","triplet_id":"x%2Fpackstack~master~Id2a3c27d8afb4a1d0e852a70e5a9ccd07e485399","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-02-10 11:47:05.000000000","reason":"Alfredo Moralejo replied on the change"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-02-10 12:50:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2a3c27d8afb4a1d0e852a70e5a9ccd07e485399","subject":"Ignore facter error when fetching repo","status":"MERGED","created":"2022-02-03 15:36:07.000000000","updated":"2022-02-10 12:50:43.000000000","submitted":"2022-02-10 12:50:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827697","meta_rev_id":"3ec283ffc364209cdcbcb4b724e11d9fe10c6100","_number":827697,"virtual_id_number":827697,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-10 12:50:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-02-10 11:47:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-02-10 11:47:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 16:14:31.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2022-02-03 16:42:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8da8ab845c8299bbb72c9e969e5069be3c61c50e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-02-03 15:36:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32f84ffa66ec02ff6d3b6d19a44c9ac2c5ce4321","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-02-03 16:03:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72d4882660ad3d49744c237803684c66144a2c9f","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-03 16:14:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73ce2464a327feeccc55fcfe76a38d493dcceade","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-03 16:15:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51f8572d66744b7811de1ced46825708255f85cc","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-02-03 16:27:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86d7a82d1b2cf2d3c1fb4a9e86ec41bac8114a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 16:42:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd74336563ca46ba8c8e5b681bff2594 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dacfcf878af461f8960599210f53158 : SUCCESS in 5m 22s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/32d183d5031a411b904f68e36b4a11f1 : SUCCESS in 57m 12s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1bc38d371f4446e08ac2f531e27d53ac : SUCCESS in 47m 22s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a6e03be0bd374b509c0410d04e6f6764 : SUCCESS in 43m 04s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/51d380550a684a0aa618af3755ab942d : SUCCESS in 56m 53s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4eae56f67bb14b91bc33fe884c9f1640 : SUCCESS in 41m 15s (non-voting)\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/12c06de8389142e6991e2e8009d3208a : SUCCESS in 41m 40s (non-voting)\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2c51688ec6af4b2e8dc94101dd151fd2 : SUCCESS in 51m 16s (non-voting)\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7eaae8b7d566451b8ebc61e63d63f69a : SUCCESS in 58m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d9a1ce99344fb76090fb03fc0307a7048fb6fed9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-02-04 10:47:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"69e59653afccb573f6b665c27f828e440656bdeb","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-02-04 10:47:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a538a3472df045dcf427fa7f9f1146ac0bd04aff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:59:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a59cb33af94e53b4a308bbe47cda45 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd897b41deb74ca3a96677073828b6de : SUCCESS in 4m 55s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a2226d245a2c48c3a128f84157e0a2e4 : SUCCESS in 44m 03s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e161df82d8d940f381cc67ab1a71e971 : SUCCESS in 46m 16s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/270f161b06be467e9989d3466f3a09c8 : SUCCESS in 51m 21s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/aaa01c90292840dd84533728774ba211 : SUCCESS in 1h 05m 59s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6ec1ef9efa36454d926f0ebb356536c7 : SUCCESS in 45m 11s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b43c3f4b6d3e45928a561b526e4dd9c7 : SUCCESS in 34m 24s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/987de4b1612f4a74ba4a3c3db11c8b82 : SUCCESS in 37m 59s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6091f5d1b9ab4649ae72bfc83ef14124 : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7d3ab320eb50d0ffcb5d87f30250f7115bf39c3d","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-10 11:47:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5f25ded4da25c3154a35131ff69f8f070564e37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 11:48:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d86e4e0cd72566d53266757e7dde2d65ff1c5502","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 12:50:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8c2d276e7a5421aba3d5ba29f1c9709 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/302ec185a6e64d5e87bf791df56db7f9 : SUCCESS in 4m 15s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b64d833a8c0c49f9bdfd406c47c302c5 : SUCCESS in 49m 59s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/fc67655377c7412e9a0127e082b7d44a : SUCCESS in 49m 25s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/05bc83cbadc54206b6e604aed773c94b : SUCCESS in 55m 23s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4415174ad2594f568a6a1655bdd9d023 : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":2},{"id":"3ec283ffc364209cdcbcb4b724e11d9fe10c6100","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 12:50:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9bfd5b848293d9bbc5d73f2319416f3793c80303","revisions":{"1a7cdfac34cff01310c14aacaffcfdee5f051630":{"kind":"REWORK","_number":1,"created":"2022-02-03 15:36:07.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/97/827697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/97/827697/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/97/827697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/97/827697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/97/827697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/97/827697/1"}}},"commit":{"parents":[{"commit":"2f69faf24acc947c0bfe4b1892cdb2bb628eec1a","subject":"Move database connection config to ::db classes for glance and cinder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2f69faf24acc947c0bfe4b1892cdb2bb628eec1a"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-02-03 15:28:49.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-02-03 15:28:49.000000000","tz":60},"subject":"Ignore facter error when fetching repo","message":"Ignore facter error when fetching repo\n\nEven though a repo is set with skip_if_unavailable\u003dTrue,\nFacter logs \"Error:...\" and then \"Ignoring repositories...\".\nWe need to add this regexp to ignored list.\n\nNote: if repo should not be skipped if unavailable, dnf\nwould have raised error before Facter run.\n\nChange-Id: Id2a3c27d8afb4a1d0e852a70e5a9ccd07e485399\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1a7cdfac34cff01310c14aacaffcfdee5f051630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1a7cdfac34cff01310c14aacaffcfdee5f051630"}]},"branch":"refs/heads/master"},"9bfd5b848293d9bbc5d73f2319416f3793c80303":{"kind":"REWORK","_number":2,"created":"2022-02-04 10:47:28.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/97/827697/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/97/827697/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/97/827697/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/97/827697/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/97/827697/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/97/827697/2"}}},"commit":{"parents":[{"commit":"2f69faf24acc947c0bfe4b1892cdb2bb628eec1a","subject":"Move database connection config to ::db classes for glance and cinder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2f69faf24acc947c0bfe4b1892cdb2bb628eec1a"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-02-03 15:28:49.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-02-04 10:47:21.000000000","tz":60},"subject":"Ignore facter error when fetching repo","message":"Ignore facter error when fetching repo\n\nEven though a repo is set with skip_if_unavailable\u003dTrue,\nFacter logs \"Error:...\" and then \"Ignoring repositories...\".\nWe need to add this regexp to ignored list.\n\nNote: if repo should not be skipped if unavailable, dnf\nwould have raised error before Facter run.\n\nChange-Id: Id2a3c27d8afb4a1d0e852a70e5a9ccd07e485399\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9bfd5b848293d9bbc5d73f2319416f3793c80303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9bfd5b848293d9bbc5d73f2319416f3793c80303"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
