)]}'
{"id":"x%2Fpackstack~828251","triplet_id":"x%2Fpackstack~master~Ib0e98d5847e2b589887ab014279bff4bdc650506","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-02-10 10:46:43.000000000","reason":"Change was submitted"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-02-10 10:46:43.000000000","reason":"Change was submitted"},"13294":{"account":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"last_update":"2022-02-10 09:34:32.000000000","reason":"Javier Peña replied on the change"}},"hashtags":[],"change_id":"Ib0e98d5847e2b589887ab014279bff4bdc650506","subject":"Move database config parameters to separated ::db classes","status":"MERGED","created":"2022-02-08 08:44:53.000000000","updated":"2022-02-10 10:46:43.000000000","submitted":"2022-02-10 10:46:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828251","meta_rev_id":"6c178643a953d4c51a8a4648909220c01059f497","_number":828251,"virtual_id_number":828251,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-10 10:46:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"recommended":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":1,"date":"2022-02-08 09:52:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-10 09:34:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-10 09:37:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 09:52:47.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2022-02-08 10:00:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-08 10:16:49.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2022-02-08 10:16:49.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"abc925e49bb9edff42ca3037b27cffb3d2906a0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-08 08:44:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"240de4246b77f3007590677038ed610819b649e2","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-02-08 09:52:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1267ec79d8a241ed6fd92f8f7a5e0ddd161a67ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 10:00:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/631c65fcf09f4a51916ea95c0d8c534d : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be48537e8174462994bdca5f9e9b02ae : SUCCESS in 4m 39s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/74fe64ce7ee7449ab0252fc9636f737a : SUCCESS in 48m 57s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4e9073bca996470a95ac941cc72447b7 : SUCCESS in 55m 16s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/901d0147a39c473bb21fe67ccfb4d715 : SUCCESS in 58m 32s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7ed6e5eef05b45628187bfcfae3f907f : SUCCESS in 1h 09m 05s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1fa65c7be5b149f59d6d996eeba554a1 : SUCCESS in 44m 19s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/70ff198497f14611a55af18a1edb2248 : SUCCESS in 43m 52s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/954481e11f8449e391e996822de0ab19 : SUCCESS in 46m 44s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/85d52fa3045640cc9fa2f8c238ae81d6 : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9f0c9424cd79392f9a16667a98b63c6828fa423e","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2022-02-10 09:34:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a1b57a4d557dad1dfaf93de9a09061d089f55ed","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2022-02-10 09:37:26.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5719712b3ce20b37ec4d4f28c6efd78942c7a884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 09:38:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"286cc4ce1a1b6706eca86f2e34e3aa8ad2fb6530","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 10:46:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e097e0c2ba4c089df5ffc53098f85c : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61cef0f704ce47bfb5cdd77e78476447 : SUCCESS in 3m 52s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/435e83728d8e42ecaeac68f5698e2518 : SUCCESS in 1h 00m 51s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/dac2782457924b6083023f12ad1a9a51 : SUCCESS in 56m 57s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/4e7c279fd3274dc9abdd93d108b10216 : SUCCESS in 1h 01m 36s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3431dcb5bec14029971d004d9f2378a7 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6c178643a953d4c51a8a4648909220c01059f497","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 10:46:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5","revisions":{"3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5":{"kind":"REWORK","_number":1,"created":"2022-02-08 08:44:53.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/51/828251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/51/828251/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/51/828251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/51/828251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/51/828251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/51/828251/1"}}},"commit":{"parents":[{"commit":"500e6e6de934e4b9dc80d8c3feb0585aaa9c6a2e","subject":"Use puppet-passed-ci repos for CS9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/500e6e6de934e4b9dc80d8c3feb0585aaa9c6a2e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-08 08:43:11.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-08 08:43:11.000000000","tz":60},"subject":"Move database config parameters to separated ::db classes","message":"Move database config parameters to separated ::db classes\n\nThe parameters have been removed for heat, ironic, keysonte, manila and\nsahara [1]. This patch is moving database_connection to the separated\n::db classed for all of them.\n\n[1] https://review.opendev.org/q/topic:cleanup-db-params\n\nChange-Id: Ib0e98d5847e2b589887ab014279bff4bdc650506\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
