)]}'
{"id":"x%2Fpackstack~828269","triplet_id":"x%2Fpackstack~master~I728bfcb6bc7c9c37bdf74dc2e357eb87696bc523","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-15 15:04:12.000000000","reason":"Change was submitted"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-02-15 14:07:59.000000000","reason":"Alfredo Moralejo replied on the change"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-02-15 11:13:53.000000000","reason":"Joel Capitao replied on the change"},"13294":{"account":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"last_update":"2022-02-11 14:09:05.000000000","reason":"Javier Peña replied on the change"}},"hashtags":[],"change_id":"I728bfcb6bc7c9c37bdf74dc2e357eb87696bc523","subject":"Remove usage of deprecated parameters","status":"MERGED","created":"2022-02-08 10:17:05.000000000","updated":"2022-02-15 15:04:12.000000000","submitted":"2022-02-15 15:04:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828269","meta_rev_id":"fc68d00e2aafc5631adfec5619d1171a6a9cd81c","_number":828269,"virtual_id_number":828269,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 15:04:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"recommended":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"date":"2022-02-15 14:08:04.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-15 11:13:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":2,"date":"2022-02-11 14:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-02-15 14:07:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 11:13:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 11:49:24.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2022-02-10 11:49:24.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2022-02-10 11:49:24.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"9106ad41576ebd17593a233fa752bbb4692b2be9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-08 10:17:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65841280ac69d32885aeec0dd187c30a2290aebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 11:13:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/336bb6dc08054225a7f04f63b9a67c32 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/614097129a4b4894940e694c90073a5b : SUCCESS in 4m 06s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/aa6a1f5f0cf748f487689d0ff8677d99 : SUCCESS in 40m 38s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/00be3bc96482493cb08a7128a9425d37 : FAILURE in 13m 16s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d2b7df8b467e4b44ad2af10f67b69c59 : SUCCESS in 45m 18s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1823523616c84129816a665714cf03b6 : FAILURE in 14m 55s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/48721d75ae0647bdac457ce3c1236c45 : SUCCESS in 43m 06s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1260f669615343d8abb8263adb9a6c73 : FAILURE in 20m 39s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ff1c155c75ac46309868fa31f21c3acd : SUCCESS in 46m 54s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1771a41ae1c54a63a42e50a646f86372 : FAILURE in 20m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ad7af9871eab4a6e949fae442be3bf3bd5dcea65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-10 09:37:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"70254bdea2441aeebf50629a4685b74c57c1dca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 10:48:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2833fbff22f4623a7ed6e832275858c : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8875ea9790d843e88394a0c106239ad1 : SUCCESS in 4m 47s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b493fb53da764f56839ddcd587fb441d : SUCCESS in 50m 36s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d2bbe0c27f4a426c8b459efefccc6f82 : SUCCESS in 47m 13s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/53ef065a18244bfaa3cd707120bd65f1 : SUCCESS in 52m 43s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ce744e9cb00146d7a06b80bd45e98a82 : SUCCESS in 1h 04m 47s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/82990854ab204b8a856c625eaee48a42 : SUCCESS in 40m 50s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d91bae0246944ba3bdd634ea56643b58 : SUCCESS in 45m 51s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/41a9aabf98d042549e842670f418ae0b : SUCCESS in 47m 39s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/30a666856094407fafb8dd828172f0b2 : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":2},{"id":"2af4cc3a847a0b38a885110b5f8a248d2e5831a8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:13:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9db62283dc13841662c7b86f6fd256c376f81178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:28:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c1fa54d60f6fe061a1436cd3a98d999e20b8a331","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:29:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79fd9ea4a6889967c5fc32b273f257cc725eabbf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:30:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"20681d9ab2dc599f58c700bbe62129d2c9eb62d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 13:35:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/413d94a56cbe435b85b3676e4688618e : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfe4d6f90afd4e2099e5fef940f71b74 : SUCCESS in 3m 49s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f0f27d43d9ea4e46a3b7caa9623373a3 : SUCCESS in 49m 14s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5dc160fbb3c241caba9280aafaa223e3 : FAILURE in 19m 28s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f43d022e4d8d4885a1e3fa4fef5a47fd : SUCCESS in 1h 01m 07s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/75b4ceb8606a4980a08e62caeeac1a9d : FAILURE in 27m 25s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c42179f172394881b2d851f96c48f4ad : SUCCESS in 46m 41s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/dda84b1dd21249d882b3b47099e0a72f : FAILURE in 20m 31s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9fe2b8761e4c42b0aa07feb66b74f517 : SUCCESS in 47m 11s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/02e066c42dae44fba842c988b50076c6 : FAILURE in 15m 51s","accounts_in_message":[],"_revision_number":3},{"id":"dec0c07da0578180105d77128cfac945d04c3ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 13:50:31.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"65fbe1a070cbed6865ad582f39388661b285ecf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 14:55:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4232d4d20ca241e2a8fdeeb04fbfc85d : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58b1371596a24efa98ea54e7343ab6fe : SUCCESS in 4m 05s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/efc58a412a604d47a8b98f6d89c0a27c : SUCCESS in 43m 21s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bafa8e4d9f7247049daad6237c8fca5d : SUCCESS in 51m 46s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b6697d6a267e49e2bfc4c199a99e5e81 : SUCCESS in 54m 19s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7eb4a67fda8b4d7fa8e90fa5f7994fb4 : SUCCESS in 58m 48s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/186320da53a9430183d7422f2d036384 : SUCCESS in 45m 56s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/9b49f6740dbf4f0186247d7c600c48f5 : SUCCESS in 43m 04s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/824f5c95c6fc413a8e7bd117af95c8ab : SUCCESS in 47m 36s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/fab7618b7d7a4849b3eb0fb98e5bb95c : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":4},{"id":"92ad4efac6976545290d7990a1bcca2184d546cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-11 09:24:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"79fd63d5634c178db7cd611b5e27188add2c04b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 10:29:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b76deb66c04459aa0e01a036579f694 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d5ad96b51944c058ebfdbdc9272f3be : SUCCESS in 4m 05s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c96181768ac94d11a0a1557a55997364 : SUCCESS in 47m 39s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0970a439d55e40a8975035c5fe7b75b1 : FAILURE in 20m 24s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ba351df909b24726849a09f3c1566fa5 : SUCCESS in 43m 36s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/50eb028181c541358df9b501b6a94484 : FAILURE in 22m 28s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c4bd99a4ebcb4d359b49c90b7af145f2 : SUCCESS in 43m 26s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/fac7521973f245b681aa9d27e42e364a : FAILURE in 18m 49s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/01a845f9db1b4d2bbad4c48ecd765ba9 : SUCCESS in 37m 45s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/aece524fbe1245a3a5957dbadfbfd10c : FAILURE in 22m 48s","accounts_in_message":[],"_revision_number":5},{"id":"2c0d8d6486e66063b3408d53423fe81685ab792a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-11 12:28:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"167cff416d3c7dd3505b0a0744c9cc83be1fe6e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 13:50:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0040bbe183bb4926a96d0c1ba97a2ad3 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b979dd38bb640378fbaa750eb4bbe6b : SUCCESS in 4m 09s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f169504174a94ee18b5755ac3a503b4e : SUCCESS in 54m 56s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5946edae359f4db49e01e13879c7323e : SUCCESS in 58m 58s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/10fe6378003a48a4ac9eae54a86211e7 : SUCCESS in 47m 56s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5d6157e6eeaf45399906ad9debc09c60 : SUCCESS in 1h 15m 44s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/bf512cf6553b4c51962061c0e3706b25 : SUCCESS in 42m 42s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/713847bcd0d948df9ed69624c95e0bfd : SUCCESS in 35m 34s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0ba06ec97d064d6d8a65887360922ba3 : SUCCESS in 48m 28s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2f9e3a569f2d44958552553a814992a3 : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":6},{"id":"3a95f965d4a662f67ea22eac55e488ed7d2d6532","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2022-02-11 14:09:05.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1bc5febd670ba661266abd38551cd34659dbef2c","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-02-15 11:13:53.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"25ed01dcf300dec20ce2897f02d252437cf5dc79","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-15 14:07:59.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"46c93bb6340ee1f044da920e93383e98c3e8579a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-15 14:08:04.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"057efe329b6abe0b9a2ce6d41e3bb22d8602f3e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 14:09:02.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"78ea5a1df8e366212324757f2bc2397cccdf95a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 15:04:12.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43d5d679efb04ac7b6e3d8e5686bc87c : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa21b614ef074b88b0294190f6c2c382 : SUCCESS in 4m 34s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/19cc8ea1157e48a8907a50a358eb2d58 : SUCCESS in 37m 44s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ac7913209406401fbbeccbbefabb7e66 : SUCCESS in 39m 53s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0345abddd2cd4065a20ca60d42852d10 : SUCCESS in 42m 32s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/91c821dccb4c46f08a9bde42218fcf3c : SUCCESS in 48m 48s","accounts_in_message":[],"_revision_number":6},{"id":"fc68d00e2aafc5631adfec5619d1171a6a9cd81c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 15:04:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e8dcebc1d726f69e725317dcefa070ed267213da","revisions":{"dff1f51db3110a09a63e50aa8dcb3167a860ca53":{"kind":"REWORK","_number":1,"created":"2022-02-08 10:17:05.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/69/828269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/69/828269/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/69/828269/1"}}},"commit":{"parents":[{"commit":"3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5","subject":"Move database config parameters to separated ::db classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-08 09:01:08.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-08 09:01:08.000000000","tz":60},"subject":"Remove usage of deprecated parameters","message":"Remove usage of deprecated parameters\n\nThis patch is removing usage of several deprecated parameters:\n\n- keystone_v2: https://review.opendev.org/507570\n- allow_versions: https://review.opendev.org/817350\n- coordination_url: https://review.opendev.org/792509\n- allow_overlaping_ips: https://review.opendev.org/817199\n\nChange-Id: I728bfcb6bc7c9c37bdf74dc2e357eb87696bc523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/dff1f51db3110a09a63e50aa8dcb3167a860ca53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/dff1f51db3110a09a63e50aa8dcb3167a860ca53"}]},"branch":"refs/heads/master"},"a3bb0e2d4b6ce9880864257d7122aae31b2781e0":{"kind":"REWORK","_number":2,"created":"2022-02-10 09:37:56.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/69/828269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/69/828269/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/69/828269/2"}}},"commit":{"parents":[{"commit":"3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5","subject":"Move database config parameters to separated ::db classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-08 09:01:08.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-10 09:37:42.000000000","tz":60},"subject":"Remove usage of deprecated parameters","message":"Remove usage of deprecated parameters\n\nThis patch is removing usage of several deprecated parameters:\n\n- keystone_v2: https://review.opendev.org/507570\n- allow_versions: https://review.opendev.org/817350\n- coordination_url: https://review.opendev.org/792509\n- allow_overlaping_ips: https://review.opendev.org/817199\n\nChange-Id: I728bfcb6bc7c9c37bdf74dc2e357eb87696bc523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a3bb0e2d4b6ce9880864257d7122aae31b2781e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a3bb0e2d4b6ce9880864257d7122aae31b2781e0"}]},"branch":"refs/heads/master"},"8bbc424a8e468046bad2026086fc12058c55537d":{"kind":"REWORK","_number":3,"created":"2022-02-10 12:28:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/828269/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/69/828269/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/69/828269/3"}}},"commit":{"parents":[{"commit":"e50a8afad2b81120d8bc17ca9f9f1e2bd0fac7f3","subject":"Fix outdated swift pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e50a8afad2b81120d8bc17ca9f9f1e2bd0fac7f3"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-08 09:01:08.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 12:28:34.000000000","tz":540},"subject":"Remove usage of deprecated parameters","message":"Remove usage of deprecated parameters\n\nThis patch is removing usage of several deprecated parameters:\n\n- keystone_v2: https://review.opendev.org/507570\n- allow_versions: https://review.opendev.org/817350\n- coordination_url: https://review.opendev.org/792509\n- allow_overlaping_ips: https://review.opendev.org/817199\n\nChange-Id: I728bfcb6bc7c9c37bdf74dc2e357eb87696bc523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8bbc424a8e468046bad2026086fc12058c55537d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8bbc424a8e468046bad2026086fc12058c55537d"}]},"branch":"refs/heads/master"},"ae8e155c4a0762237d3a2df449f378fb8a916b6d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-10 13:50:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/828269/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/69/828269/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/69/828269/4"}}},"commit":{"parents":[{"commit":"7c97f3963973c8c8bff62253d799c9cfe98dfc10","subject":"Fix outdated swift pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7c97f3963973c8c8bff62253d799c9cfe98dfc10"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-08 09:01:08.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 13:50:22.000000000","tz":540},"subject":"Remove usage of deprecated parameters","message":"Remove usage of deprecated parameters\n\nThis patch is removing usage of several deprecated parameters:\n\n- keystone_v2: https://review.opendev.org/507570\n- allow_versions: https://review.opendev.org/817350\n- coordination_url: https://review.opendev.org/792509\n- allow_overlaping_ips: https://review.opendev.org/817199\n\nChange-Id: I728bfcb6bc7c9c37bdf74dc2e357eb87696bc523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ae8e155c4a0762237d3a2df449f378fb8a916b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ae8e155c4a0762237d3a2df449f378fb8a916b6d"}]},"branch":"refs/heads/master"},"e8d2b29c9b44c656a8b54af9861df5fecd18cfc9":{"kind":"REWORK","_number":5,"created":"2022-02-11 09:24:42.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/69/828269/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/69/828269/5","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/69/828269/5"}}},"commit":{"parents":[{"commit":"e61867cb238e81f6580bb3ac30e38118fa2c697c","subject":"Merge \"Fix outdated swift pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e61867cb238e81f6580bb3ac30e38118fa2c697c"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-08 09:01:08.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-11 09:18:10.000000000","tz":60},"subject":"Remove usage of deprecated parameters","message":"Remove usage of deprecated parameters\n\nThis patch is removing usage of several deprecated parameters:\n\n- keystone_v2: https://review.opendev.org/507570\n- allow_versions: https://review.opendev.org/817350\n- coordination_url: https://review.opendev.org/792509\n- allow_overlaping_ips: https://review.opendev.org/817199\n\nChange-Id: I728bfcb6bc7c9c37bdf74dc2e357eb87696bc523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e8d2b29c9b44c656a8b54af9861df5fecd18cfc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e8d2b29c9b44c656a8b54af9861df5fecd18cfc9"}]},"branch":"refs/heads/master"},"e8dcebc1d726f69e725317dcefa070ed267213da":{"kind":"REWORK","_number":6,"created":"2022-02-11 12:28:18.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/69/828269/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/69/828269/6","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/69/828269/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/69/828269/6"}}},"commit":{"parents":[{"commit":"e61867cb238e81f6580bb3ac30e38118fa2c697c","subject":"Merge \"Fix outdated swift pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e61867cb238e81f6580bb3ac30e38118fa2c697c"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-08 09:01:08.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-02-11 12:28:10.000000000","tz":60},"subject":"Remove usage of deprecated parameters","message":"Remove usage of deprecated parameters\n\nThis patch is removing usage of several deprecated parameters:\n\n- keystone_v2: https://review.opendev.org/507570\n- allow_versions: https://review.opendev.org/817350\n- coordination_url: https://review.opendev.org/792509\n- allow_overlaping_ips: https://review.opendev.org/817199\n\nChange-Id: I728bfcb6bc7c9c37bdf74dc2e357eb87696bc523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e8dcebc1d726f69e725317dcefa070ed267213da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e8dcebc1d726f69e725317dcefa070ed267213da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
