)]}'
{"id":"x%2Fpackstack~828706","triplet_id":"x%2Fpackstack~master~I8578eb5087930147f461b3db41a183998edef796","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-10 17:06:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8578eb5087930147f461b3db41a183998edef796","subject":"Fix outdated swift pipeline","status":"MERGED","created":"2022-02-10 12:28:45.000000000","updated":"2022-02-10 17:06:27.000000000","submitted":"2022-02-10 17:06:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828706","meta_rev_id":"fe5fa59531e23f583b1b5b2374b5351daf32482d","_number":828706,"virtual_id_number":828706,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-10 17:06:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-02-10 15:44:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-02-10 15:44:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 13:27:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 14:00:07.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2022-02-10 15:44:02.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"9556482cac6d783c18a701cd61d715380b62e38f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:28:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b46a3b53efccde9e690007a4d00bfe401ed597cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 13:27:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b01c0a05ca4957a4a7af3d43ab6145 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29410f36687a400281c5bae85de70f6d : SUCCESS in 4m 25s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e8aa9737907d48278eea0e5543bb27bd : SUCCESS in 44m 59s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3453aa5321c943f3a115728683cec896 : FAILURE in 11m 56s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/961abf4651da4e9c990c649ead4bf174 : SUCCESS in 43m 33s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/46371812b4fe42879cd0eff806826429 : FAILURE in 14m 21s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/699270de4ac94144a214f459b34c59a6 : SUCCESS in 37m 44s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/dec11635e6d84a08932ae04b582bcab4 : FAILURE in 11m 11s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2ee40d5433c14fe78b677bc40e5674c9 : SUCCESS in 53m 15s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/aa797a4142bd4b779fa6e482cea0cd51 : FAILURE in 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"fab2d47550187edef18452ceb7c273b97e41a47d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 13:50:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0d84ca094f77dd5d714f65e09ad576ea8a03b59e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-10 14:00:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c56a7430c781f4c8f74a507288582096e944ed7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 15:03:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d24509c9da86415294d8faa71057bc81 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22d266038607417a8d9bab1c497e81a8 : SUCCESS in 4m 14s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/3133e5d285bd4aa18606a2713e78e77d : SUCCESS in 40m 25s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/452e86bd4a5542d4910afe3f3f562354 : SUCCESS in 41m 10s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/86ff43c751064a00b0fec81ca52ad021 : SUCCESS in 54m 13s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/10a7abcaa81a44dbb76fc55f62f72b9f : SUCCESS in 59m 10s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/00b8332362494969a89d4eb8749e7ebf : SUCCESS in 49m 48s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/a49a6461f4714d69ba9a56cfc08689d6 : SUCCESS in 41m 25s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b6d9f8cb31c146c8b0da592a3f67975e : SUCCESS in 48m 25s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0c5c159816704411905d30ad5cfcaf6e : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":2},{"id":"989827887423718341f06fd7d0cfee0e732ce16c","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-02-10 15:44:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b8869e7348aa40ccbf1146e4466f931ef1391034","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 15:45:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"07c4e3a464d840f595f2752c127b27f280cd7f71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 17:06:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d82acffdb194ed39c63c9c52a8f75e3 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d254cd7a89cb4a7381826890ed21e95d : SUCCESS in 4m 33s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6ffcfe2b8d1445df9ff1f791eddd838a : SUCCESS in 53m 38s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7b8b9e4bfa0e408598898e136e9c6102 : SUCCESS in 53m 49s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a17f80ebefb54a769106eb6aa02d2fcf : SUCCESS in 1h 15m 28s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/700941ba2f0046cdab5d8a5a6ab93430 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":2},{"id":"fe5fa59531e23f583b1b5b2374b5351daf32482d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 17:06:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7c97f3963973c8c8bff62253d799c9cfe98dfc10","revisions":{"e50a8afad2b81120d8bc17ca9f9f1e2bd0fac7f3":{"kind":"REWORK","_number":1,"created":"2022-02-10 12:28:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/828706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/06/828706/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/06/828706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/06/828706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/06/828706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/06/828706/1"}}},"commit":{"parents":[{"commit":"3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5","subject":"Move database config parameters to separated ::db classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 12:22:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 12:27:33.000000000","tz":540},"subject":"Fix outdated swift pipeline","message":"Fix outdated swift pipeline\n\nThis change fixes the outdated swift pipeline, following the recent\ndefault pipeline defined in example proxy-server.conf.\n\n- Add missing fundamental middleware\n  (gatekeeper, proxy-logging)\n\n- Add middleware to provide basic functionalities\n  (listing_fomrats, copy, dlo)\n\n- tempurl middleware should be placed BEFORE auth middleware\n\n- Put bulk middleware before ratelimit following the example file.\n\nChange-Id: I8578eb5087930147f461b3db41a183998edef796\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e50a8afad2b81120d8bc17ca9f9f1e2bd0fac7f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e50a8afad2b81120d8bc17ca9f9f1e2bd0fac7f3"}]},"branch":"refs/heads/master"},"7c97f3963973c8c8bff62253d799c9cfe98dfc10":{"kind":"REWORK","_number":2,"created":"2022-02-10 13:50:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/828706/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/06/828706/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/06/828706/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/06/828706/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/06/828706/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/06/828706/2"}}},"commit":{"parents":[{"commit":"3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5","subject":"Move database config parameters to separated ::db classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3d4a38a9fc0a3f55bd57ba9112cc999bb44582e5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 12:22:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 13:50:17.000000000","tz":540},"subject":"Fix outdated swift pipeline","message":"Fix outdated swift pipeline\n\nThis change fixes the outdated swift pipeline, following the recent\ndefault pipeline defined in example proxy-server.conf.\n\n- Add missing fundamental middleware\n  (gatekeeper, proxy-logging)\n\n- Add middleware to provide basic functionalities\n  (listing_fomrats, copy, dlo)\n\n- tempurl middleware should be placed BEFORE auth middleware\n\n- Put bulk middleware before ratelimit following the example file.\n\nChange-Id: I8578eb5087930147f461b3db41a183998edef796\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7c97f3963973c8c8bff62253d799c9cfe98dfc10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7c97f3963973c8c8bff62253d799c9cfe98dfc10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
