)]}'
{"id":"x%2Fpackstack~832726","triplet_id":"x%2Fpackstack~stable%2Fxena~If6270b7f49529382c5b8147449eb631c6df420aa","project":"x/packstack","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-03-16 04:50:18.000000000","reason":"yatin replied on the change"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-03-16 06:22:50.000000000","reason":"Change was submitted"},"13294":{"account":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"last_update":"2022-03-15 16:52:28.000000000","reason":"Javier Peña replied on the change"}},"hashtags":[],"change_id":"If6270b7f49529382c5b8147449eb631c6df420aa","subject":"Use ecdsa key in CentOS 9 Stream","status":"MERGED","created":"2022-03-09 13:18:48.000000000","updated":"2022-03-16 06:22:50.000000000","submitted":"2022-03-16 06:22:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832726","cherry_pick_of_change":831795,"cherry_pick_of_patch_set":2,"meta_rev_id":"bc5da529c17344995e20330544fbe8a28b1d561e","_number":832726,"virtual_id_number":832726,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-16 06:22:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-03-16 04:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-15 16:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-15 17:12:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-09 13:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 18:19:50.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2022-03-10 18:19:50.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-03-15 23:36:20.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"1032a709d5cf13435249633489342728383886f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-03-09 13:18:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a73ceb497fd2e0a103487023f43fe0e0e38778b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 13:48:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd98fca56ba24eaf8d9e8e658d66b119 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9efb3d5089684bf3804e866c775a3b36 : SUCCESS in 4m 59s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/02d5b28e44f84664a78f6444c665e529 : FAILURE in 24m 16s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d2d34bef1bd44a82ab261f20e84cf7f6 : FAILURE in 20m 34s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/28b74c47f104481ea5ee9e51a1bc8ee2 : FAILURE in 20m 34s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/29ed70b59cf1473fa49dd6a474bf013c : FAILURE in 21m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7d3a07e0ca4b30535113c560415e2d133661bf46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-03-10 16:58:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d6d1f2cc46d8255f0895b2a43f68ae6fa6586d47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 18:06:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d757e96106e54b94b7a9bc836cbd1c01 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2122d8f735d549f19d81ae9bc06aa4e4 : SUCCESS in 5m 44s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4ce5f2a57a344cc7b77156196ceec0e2 : SUCCESS in 38m 42s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cf39ab3e8c9c45dbb2d43f4ac84ba7e3 : SUCCESS in 40m 54s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f67b338ae4d741e68ef61d9de2fec6f8 : SUCCESS in 51m 53s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/42af682c8e2c4204a78c4b139169a9c7 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":2},{"id":"68a37b286cc089366343e5914ee75bce248ca316","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2022-03-15 16:52:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8d1e7d4967cbd11045eaf9d5d64b2b8d072638ad","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2022-03-15 17:12:45.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"80015749399f00370290e5bf2c203b6d73d0913a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:14:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c6cacd6359deda8bc0119d9bac5c257d5bc7ff92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 18:04:52.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"6861f3da24c0c7784c5c639ca627efda478c9a48","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-15 23:36:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"493062a588d985e299d7f8f8c5ed05c519cbd3b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 00:40:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8b2c7b1c684c2b9267e33a755d550b : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62b9ed881d4c4c7eb322d0e0234ec8db : SUCCESS in 6m 37s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2f53c34090d24d3bbab568e0c4e35ae6 : SUCCESS in 44m 46s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/669ff47adb174d90ba9933cf87f05ecf : SUCCESS in 44m 11s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e8e6491e48584c37b1e2eaeb8d40a7f7 : SUCCESS in 59m 26s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/425a6fb40a8a449586fbe1cbe0e795c6 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":2},{"id":"641fd27445bae5d27043a9c123d32bd947dc0d38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 00:41:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c58a2658417467268c6dff30f34c08c0249cfed3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 02:46:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6f4ffc595044ef82223b78a30426fb : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ed400a0972e40c9ae2364270d1178a0 : SUCCESS in 5m 51s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/02e03e012b8840d88b8d870ab07dc54d : TIMED_OUT in 2h 03m 39s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4b9ce9e8090e4683be512aae83b9e253 : SUCCESS in 38m 06s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/07c06e4e611345ccb986e8fd3c266090 : SUCCESS in 42m 47s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/05bd0346a64c4b5f8dd19dc3c7444903 : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":2},{"id":"0f32968a8e2a190eba5b1a14ff5a89dd8e62e717","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-16 04:50:18.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"107ce333e946990d17bb883011c974d85b62010c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 05:38:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11bfbf45a7714dc8aeda1d4906aab4fc : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a2b1f860df343d7adb559b1ebf6bd90 : SUCCESS in 5m 07s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/213a02640ae54388ae92cba69eee6cf2 : SUCCESS in 35m 39s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bf0d2b912e8143f99a26d2a784a8b4b5 : SUCCESS in 37m 26s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8f05b3c43cf74e5e9580f746e9778a84 : SUCCESS in 40m 25s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a1dfd56cdb7246c5a5a0e1a790cffc9c : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":2},{"id":"913942c89296b7c9a9023ac61c6b231340ff9c0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 05:39:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f2ad2c7f1efd2fc3f5da6e7f8db06f432cda5835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 06:22:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9157af8bebbe4212ba37d3ba7c892664 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32373e8625cd425cab479b7f5d9b9b9b : SUCCESS in 6m 08s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/fa87861a49c646cfb6013ec29d279db7 : SUCCESS in 35m 46s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6dba091edb6141f8b714d3623f48c0cd : SUCCESS in 37m 55s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7d373bdf384846ccb85285d4f439d2ac : SUCCESS in 39m 10s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/42a1f6ace68548109ca2b3364fea62b1 : SUCCESS in 42m 57s","accounts_in_message":[],"_revision_number":2},{"id":"bc5da529c17344995e20330544fbe8a28b1d561e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 06:22:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b13d33ecb80f85e1b8d95d325c2e93795cd5229e","revisions":{"c4653be560799d9f51d3d54231666b36f751e7f3":{"kind":"REWORK","_number":1,"created":"2022-03-09 13:18:48.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/26/832726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/26/832726/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/26/832726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/26/832726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/26/832726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/26/832726/1"}}},"commit":{"parents":[{"commit":"ee60ad92a0c43909f4ede85eb17f81f45a12fa26","subject":"Aodh: Use the new coordination class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ee60ad92a0c43909f4ede85eb17f81f45a12fa26"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-03-03 15:25:15.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-03-09 13:18:48.000000000","tz":0},"subject":"Use ecdsa key in CentOS 9 Stream","message":"Use ecdsa key in CentOS 9 Stream\n\n... as rsa key no longer works with the latest openssl in CentOS 9\nStream repo.\n\nCo-Authored-By: Takashi Kajinami \u003ctkajinam@redhat.com\u003e\n\nRelated-Bug: #1962507\nChange-Id: If6270b7f49529382c5b8147449eb631c6df420aa\n(cherry picked from commit 6adc4339dcd2c3f0bf5b563718ce2e20bb0c4359)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c4653be560799d9f51d3d54231666b36f751e7f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c4653be560799d9f51d3d54231666b36f751e7f3"}]},"branch":"refs/heads/stable/xena"},"b13d33ecb80f85e1b8d95d325c2e93795cd5229e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-10 16:58:32.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/26/832726/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/26/832726/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/26/832726/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/26/832726/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/26/832726/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/26/832726/2"}}},"commit":{"parents":[{"commit":"4e65f539d31967b1251919283c1235255576d4ab","subject":"Use Puppet OpenStack modules from stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4e65f539d31967b1251919283c1235255576d4ab"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-03-03 15:25:15.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-03-10 16:58:32.000000000","tz":0},"subject":"Use ecdsa key in CentOS 9 Stream","message":"Use ecdsa key in CentOS 9 Stream\n\n... as rsa key no longer works with the latest openssl in CentOS 9\nStream repo.\n\nCo-Authored-By: Takashi Kajinami \u003ctkajinam@redhat.com\u003e\n\nRelated-Bug: #1962507\nChange-Id: If6270b7f49529382c5b8147449eb631c6df420aa\n(cherry picked from commit 6adc4339dcd2c3f0bf5b563718ce2e20bb0c4359)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b13d33ecb80f85e1b8d95d325c2e93795cd5229e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b13d33ecb80f85e1b8d95d325c2e93795cd5229e"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
