)]}'
{"id":"x%2Fpackstack~832999","triplet_id":"x%2Fpackstack~master~I0b2df8c5473692786f8f56c3c52ee425ba28bf8e","project":"x/packstack","branch":"master","topic":"xena","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-03-10 13:52:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0b2df8c5473692786f8f56c3c52ee425ba28bf8e","subject":"Use Puppet OpenStack modules from stable/xena","status":"ABANDONED","created":"2022-03-10 08:38:33.000000000","updated":"2022-03-10 13:52:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b6e2943ab69540b400f7845269cc859939426022","_number":832999,"virtual_id_number":832999,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-10 13:52:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 09:44:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87af48af4324103cdf937327a62bd4ada3975539","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-03-10 08:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"642176ef41cdc649a32c68777ae7882154fd1569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 09:44:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8170e6744ab4b949be9b977d1b39740 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/840528ea059c4ebfaceeff3c31dae8e4 : SUCCESS in 5m 58s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/de7cfccbca424f9b80a1fd0ea18dabb1 : FAILURE in 14m 13s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/be7c610fa2b94072a24854f3fd9e7957 : FAILURE in 15m 53s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5d952d42731d41eabd6aae9cdf1c482f : FAILURE in 13m 34s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/91768a0e4b654dd29952bc7d88856d62 : FAILURE in 17m 15s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/189e66bd0f4d4e1a84f590d05649deaa : FAILURE in 15m 46s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/28f021d3ab364352925a189371a350c9 : FAILURE in 11m 52s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d4e5dbadca9a4f729dfd8b818e98040b : FAILURE in 13m 58s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ccdfda90d75a4d4ea83ca1dc738e5020 : FAILURE in 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"84da64a37735ef73ddcb138028cfe99dd145a294","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-03-10 11:52:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce9816d97428d4b77d05a0726b27d1886ba46dce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 12:19:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/372d0a0129884cb6b111c868598f25dd : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abda7cb2ec4c4caaba956f458ec35d80 : SUCCESS in 5m 50s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1a8c934558a9410cacb07a012636d722 : FAILURE in 15m 36s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/999f1c682f744aaca8cff8bfdc5a5af0 : FAILURE in 15m 11s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f24eb284e4044e1ba7c617fe0ab1dd0d : FAILURE in 14m 57s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/df679d8c9ce941dba9b755fcb5389383 : FAILURE in 16m 28s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6525132945224e10ac51498cff9c70dd : FAILURE in 12m 15s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/56ed0a4be65e409ebef6c98b7e81478f : FAILURE in 11m 48s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7c745044cba14774aa258941ca1a6701 : FAILURE in 12m 05s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2dea4ce9cca2415e8c0607f3f7303335 : FAILURE in 13m 19s","accounts_in_message":[],"_revision_number":2},{"id":"65febb11310661d300e5e4373c3bf207a6e6165b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-03-10 13:52:34.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fe856585259422e4513155d67b55d9cb147fd573","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-03-10 13:52:51.000000000","message":"Abandoned\n\nwrong branch","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b1dcca1707ba8a116edf24da4e8469ccd159770d","revisions":{"ce489965509f108229e4a09323b7e6822cf26011":{"kind":"REWORK","_number":1,"created":"2022-03-10 08:38:33.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/99/832999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/99/832999/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/99/832999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/99/832999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/99/832999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/99/832999/1"}}},"commit":{"parents":[{"commit":"ee60ad92a0c43909f4ede85eb17f81f45a12fa26","subject":"Aodh: Use the new coordination class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ee60ad92a0c43909f4ede85eb17f81f45a12fa26"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-03-10 08:37:03.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-03-10 08:38:26.000000000","tz":60},"subject":"Use Puppet OpenStack modules from stable/xena","message":"Use Puppet OpenStack modules from stable/xena\n\nWhen testing packstack in stable/xena branch we should use the same\nbranch for the modules provided from OpenStack projects.\n\nChange-Id: I0b2df8c5473692786f8f56c3c52ee425ba28bf8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ce489965509f108229e4a09323b7e6822cf26011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ce489965509f108229e4a09323b7e6822cf26011"}]},"branch":"refs/heads/master"},"6210381dbeb161a84d16411ea416161975ccd910":{"kind":"REWORK","_number":2,"created":"2022-03-10 11:52:35.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/99/832999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/99/832999/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/99/832999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/99/832999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/99/832999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/99/832999/2"}}},"commit":{"parents":[{"commit":"ee60ad92a0c43909f4ede85eb17f81f45a12fa26","subject":"Aodh: Use the new coordination class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ee60ad92a0c43909f4ede85eb17f81f45a12fa26"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-03-10 08:37:03.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-03-10 11:52:09.000000000","tz":60},"subject":"Use Puppet OpenStack modules from stable/xena","message":"Use Puppet OpenStack modules from stable/xena\n\nWhen testing packstack in stable/xena branch we should use the same\nbranch for the modules provided from OpenStack projects.\n\nAlso pin non-openstack modules to the commits used in RDO Xena.\n\nChange-Id: I0b2df8c5473692786f8f56c3c52ee425ba28bf8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6210381dbeb161a84d16411ea416161975ccd910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6210381dbeb161a84d16411ea416161975ccd910"}]},"branch":"refs/heads/master"},"b1dcca1707ba8a116edf24da4e8469ccd159770d":{"kind":"NO_CHANGE","_number":3,"created":"2022-03-10 13:52:34.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/99/832999/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/99/832999/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/99/832999/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/99/832999/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/99/832999/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/99/832999/3"}}},"commit":{"parents":[{"commit":"ee60ad92a0c43909f4ede85eb17f81f45a12fa26","subject":"Aodh: Use the new coordination class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ee60ad92a0c43909f4ede85eb17f81f45a12fa26"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-03-10 08:37:03.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-03-10 13:52:13.000000000","tz":60},"subject":"Use Puppet OpenStack modules from stable/xena","message":"Use Puppet OpenStack modules from stable/xena\n\nWhen testing packstack in stable/xena branch we should use the same\nbranch for the modules provided from OpenStack projects.\n\nAlso pin non-openstack modules to the commits used in RDO Xena.\n\nChange-Id: I0b2df8c5473692786f8f56c3c52ee425ba28bf8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b1dcca1707ba8a116edf24da4e8469ccd159770d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b1dcca1707ba8a116edf24da4e8469ccd159770d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
